)]}'
{"id":"openstack%2Fos-testr~816562","triplet_id":"openstack%2Fos-testr~master~I83598d6657cd283510a24f6e9d695eb4914e1f8c","project":"openstack/os-testr","branch":"master","topic":"py39-voting","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-11-04 13:45:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83598d6657cd283510a24f6e9d695eb4914e1f8c","subject":"Test python3.9 as voting job","status":"MERGED","created":"2021-11-03 20:30:04.000000000","updated":"2021-11-04 13:55:36.000000000","submitted":"2021-11-04 13:45:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816562-py39-voting","meta_rev_id":"deb351e3da73445e878591a020b0b217644f83e2","_number":816562,"virtual_id_number":816562,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 13:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-11-04 13:05:06.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-11-04 13:05:06.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-03 20:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 13:05:06.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"e02d500ef5633f2811bdb3f6a8a7c191d2558a47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-03 20:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"979b9a6ee1f3aaf126bfc7841f7428685a176f00","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-03 20:38:41.000000000","message":"Topic set to py39-testing","accounts_in_message":[],"_revision_number":1},{"id":"88e5cadd309c4a83a85f387b2d01bc543c8c437b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 20:41:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc35d9172f4941499f1d0b097a1950f2 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b46201774746088dbb43ab20fc1da0 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671172053bb840a4a7fa1a21dea6a781 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df15ae88f3c747ab9fa2123730a96a2f : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f245e41a1344a0e9c954d6ac3b22d6c : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dcc7894622e429daa694cdda6a576a9 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ae97ebec088468b8c757d203771be4b : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"63f1bbff633d0678a987917248eec287e999cc38","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-04 00:32:48.000000000","message":"Topic py39-testing removed","accounts_in_message":[],"_revision_number":1},{"id":"71ce010d59d4f6f0acac5b7580dbf75b3505de24","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-04 00:33:03.000000000","message":"Topic set to py39-voting","accounts_in_message":[],"_revision_number":1},{"id":"b184323b13ea9c4eb24697f2b46bc27a1c2704b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-04 13:04:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e46c3c94f62b5fead19053fa8c4b9127e8d87025","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-04 13:05:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aece5f6e065d718d734d2e8400bc9a2481d8e21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 13:19:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b82f7f5d40044abaa26163c40b3f6601 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ccf55e5db314310b7d3ed84604ac74b : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2beb9b4347047cd88db0dbf2e411435 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c6632bda6e54cbf91ae1c78aeb91259 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c68ae9f921a482689e919ccecbdc5d1 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63cb41ee708f4389a8bdbac1203b83aa : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61af19652d3f4edca87bc13306ba3482 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"7b0db53215c2df4490eaac5d7da4b2a449691df0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 13:26:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3efde78c21aa3b726a3559507b43759fd5e98141","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 13:45:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a296f9f561a144f496bca08565690cfa : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2fb4cd2d934bed975d995b5f500cf1 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70911dbd0ee4f27a46771fea6ea1d91 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b45797c2e8314da9b26418b50eb024a9 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46dbc4fc15554e0bbc2a843f5c90565c : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0394ffabd779459489ca6092729768e2 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1b889a4974540e481dcd1131b655aeb : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8655611519b0072fbea10d528ebec07717ae4e29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 13:45:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"deb351e3da73445e878591a020b0b217644f83e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 13:55:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c81d81f1cbe4cf2acd2778c44fb424e : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4506fcf719990e1fc4c3fce85f2eb469b4c5a2d5","revisions":{"d94090c0954cca955aa10451813d8925d7f1de01":{"kind":"REWORK","_number":1,"created":"2021-11-03 20:30:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/62/816562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/816562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/816562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/816562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/816562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/816562/1"}}},"commit":{"parents":[{"commit":"c0710b984b90b428408fdb424823f1be77352a30","subject":"Switch testing to Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/c0710b984b90b428408fdb424823f1be77352a30"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 20:29:36.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 20:29:36.000000000","tz":-300},"subject":"Test python3.9 as voting job","message":"Test python3.9 as voting job\n\nIn Yoga testing, we are making py3.9 job to run as voting, so\nadding py3.9 job as voting.\n\nAlso, as Tempest is branchless and needs to support older\nsupported stable branches, let\u0027s test it with all the\npython version supported in those supported stable branches.\nFor that, we need to explicitly dd the py jobs instead of using\nthe template. This way, we will explicitly know that we are testing\nall supported python versions when the new release template bumps the\nmin python version in the template.\n\nChange-Id: I83598d6657cd283510a24f6e9d695eb4914e1f8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/d94090c0954cca955aa10451813d8925d7f1de01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/d94090c0954cca955aa10451813d8925d7f1de01"}]},"branch":"refs/heads/master"},"4506fcf719990e1fc4c3fce85f2eb469b4c5a2d5":{"kind":"REWORK","_number":2,"created":"2021-11-04 13:04:53.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/62/816562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/816562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/816562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/816562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/816562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/816562/2"}}},"commit":{"parents":[{"commit":"c0710b984b90b428408fdb424823f1be77352a30","subject":"Switch testing to Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/c0710b984b90b428408fdb424823f1be77352a30"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 20:29:36.000000000","tz":-300},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-11-04 13:04:53.000000000","tz":0},"subject":"Test python3.9 as voting job","message":"Test python3.9 as voting job\n\nIn Yoga testing, we are making py3.9 job to run as voting, so\nadding py3.9 job as voting.\n\nAlso, as Tempest is branchless and needs to support older\nsupported stable branches, let\u0027s test it with all the\npython version supported in those supported stable branches.\nFor that, we need to explicitly dd the py jobs instead of using\nthe template. This way, we will explicitly know that we are testing\nall supported python versions when the new release template bumps the\nmin python version in the template.\n\nChange-Id: I83598d6657cd283510a24f6e9d695eb4914e1f8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/4506fcf719990e1fc4c3fce85f2eb469b4c5a2d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/4506fcf719990e1fc4c3fce85f2eb469b4c5a2d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
