)]}'
{"id":"openstack%2Fos-testr~851301","triplet_id":"openstack%2Fos-testr~master~Iea91567941e16007750498114d279dde12571fb8","project":"openstack/os-testr","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-09-21 10:23:30.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-09-21 10:10:31.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}},"hashtags":[],"change_id":"Iea91567941e16007750498114d279dde12571fb8","subject":"remove unicode from code","status":"MERGED","created":"2022-07-28 02:09:22.000000000","updated":"2022-09-21 10:24:37.000000000","submitted":"2022-09-21 10:23:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851301","meta_rev_id":"f284ca4211d8b782de09087c8da202cdd5996341","_number":851301,"virtual_id_number":851301,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 10:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-09-21 10:10:31.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-09-21 10:10:31.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 02:27:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 22:01:51.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"95e1288d9eeb2ab4cedf303d43e5ad5a6c3037a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-07-28 02:09:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c2b6ef882aa3be07a04f5a06fc42a272ef22f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 02:27:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23690f20955841448455cbe2b9ae1eb8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99db1f81577458e8b4a94b0b1f5f03f : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e5c844c2f174ab684485fe5855628d1 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4525cefcd74423ef7dc249b81d421823877f00ec","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-09-20 22:01:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aaf7980bd493796bdb78830e0b1485b84e3baf28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-09-21 00:23:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fa48cfafec565cb00337bb1fd856a6aed8a0068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 00:36:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e521501981f740069a615a7b75d132d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06834231ecf14f298500b0879d8922ed : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3f5fb659e28481d8605fe5b7c5f6195 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3354c2068c6b4394ad18e32fd042b8f7 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9d61e8d3c74072a6e46ecb1694a7d5 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"875ea1d89025bcf2a0c7e7abc318511797239815","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-09-21 01:31:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26f5b55e6b53d25ef0f8c19c9da7b9e8bf54c649","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-09-21 10:10:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a34c81102abc3986dd953f870ef4ff6c25d0693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 10:11:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cff8154b0afa427f0302b41573540e1335ed164f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 10:23:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa66a0847c1543f4b26b828ebbac38f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51a47fd37554cf483375da85359fe6a : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cebd0567e13443e82e98796b39e8ba4 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20c413c0276847f1a19a036ad654be54 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c307b1ef0bb94c998f24640dc7a83cec : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"475d5d913b72751d56ea8556ced058e6fc25cc41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 10:23:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f284ca4211d8b782de09087c8da202cdd5996341","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 10:24:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf3528212b948ca930de66d0264bf79\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f123f2310546f3a26cfceaeec8d054 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"204fe38386a884510cdb4068ab3cb1f9bcca989b","revisions":{"a57b7bf6ab1a058b64c57812831606038a6a5e95":{"kind":"REWORK","_number":1,"created":"2022-07-28 02:09:22.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/01/851301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/01/851301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/01/851301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/01/851301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/01/851301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/01/851301/1"}}},"commit":{"parents":[{"commit":"0e7ac78c8282198bbe3706165da7dbac2a6278e5","subject":"Merge \"Update CI to use unversioned jobs template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/0e7ac78c8282198bbe3706165da7dbac2a6278e5"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-28 02:08:43.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-28 02:08:50.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Iea91567941e16007750498114d279dde12571fb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a57b7bf6ab1a058b64c57812831606038a6a5e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a57b7bf6ab1a058b64c57812831606038a6a5e95"}]},"branch":"refs/heads/master"},"204fe38386a884510cdb4068ab3cb1f9bcca989b":{"kind":"REWORK","_number":2,"created":"2022-09-21 00:23:05.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/01/851301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/01/851301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/01/851301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/01/851301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/01/851301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/01/851301/2"}}},"commit":{"parents":[{"commit":"0e7ac78c8282198bbe3706165da7dbac2a6278e5","subject":"Merge \"Update CI to use unversioned jobs template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/0e7ac78c8282198bbe3706165da7dbac2a6278e5"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-28 02:08:43.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-09-21 00:22:13.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Iea91567941e16007750498114d279dde12571fb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/204fe38386a884510cdb4068ab3cb1f9bcca989b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/204fe38386a884510cdb4068ab3cb1f9bcca989b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[]}
