)]}'
{"id":"openstack%2Fos-traits~652996","triplet_id":"openstack%2Fos-traits~master~I42237b196644bfa87701f978a1519941d447b450","project":"openstack/os-traits","branch":"master","topic":"bp/image-metadata-prefiltering","hashtags":[],"change_id":"I42237b196644bfa87701f978a1519941d447b450","subject":"add libvirt image metadata traits","status":"MERGED","created":"2019-04-16 13:43:38.000000000","updated":"2019-06-24 13:25:12.000000000","submitted":"2019-06-24 13:25:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652996-1561382712803-b95397b7","meta_rev_id":"7de024e20d5101961d82e163983125b17081393b","_number":652996,"virtual_id_number":652996,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-06-24 13:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-06-21 03:52:26.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-06-24 10:38:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-06-24 13:12:53.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 18:40:00.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-24 10:38:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-24 13:12:53.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-24 13:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f10a17148f1e8a2e1588574faa4080f7dcefca96","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-16 13:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2f1686ca7e002a8aa2fc1c6800730c063143993","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 13:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/96/652996/1/check/openstack-tox-pep8/e8546d8/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/96/652996/1/check/openstack-tox-py27/80ccabc/ : SUCCESS in 6m 49s\n- openstack-tox-docs http://logs.openstack.org/96/652996/1/check/openstack-tox-docs/3f04380/html/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/96/652996/1/check/openstack-tox-py36/5337dfe/ : SUCCESS in 10m 53s\n- openstack-tox-py37 http://logs.openstack.org/96/652996/1/check/openstack-tox-py37/8b54ac9/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4937eb35a6cabfd8506802862ab6ff1d1bbd5094","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-17 18:40:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nIn addition to the typos in the comment message, there\u0027s a been a precedent of including a URL that points to the authoritative source for a set of traits, if one exists. Is there one for these.\n\nSee https://opendev.org/openstack/os-traits/src/branch/master/os_traits/hw/cpu/amd.py#L17 for a small example","accounts_in_message":[],"_revision_number":1},{"id":"29fe88e5ed7351442bf291f16532d7b1144b4b34","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-24 14:06:13.000000000","message":"Patch Set 1:\n\nthanks ill fix the typos.\n\nas for athroitive sources i can link to the glance metadefs.\n\nill do that in then next version.","accounts_in_message":[],"_revision_number":1},{"id":"da666685a5d115203e20de7c4748d4b149e129e4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-01 19:40:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"87a0041cb003ddcf567060d38d3ba03a075d3ed8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 19:48:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/96/652996/2/check/openstack-tox-pep8/998aafe/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/96/652996/2/check/openstack-tox-py27/1c755e5/ : SUCCESS in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/96/652996/2/check/openstack-tox-docs/a8cc58a/html/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/96/652996/2/check/openstack-tox-py36/9c56f36/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/96/652996/2/check/openstack-tox-py37/07bd64c/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"cbc0cbc56e00d3a79a78b36aa309b2956d8ae558","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-01 20:58:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)\n\nI\u0027d like some explanation (in the files, not in the review) of the discrepancies between the nova and glance lists. I also think there should be links to the glance source as well as the nova source.","accounts_in_message":[],"_revision_number":2},{"id":"8fae7e1cd90eac147a8b562a5d34313ffc1415ac","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-05-01 21:24:13.000000000","message":"Patch Set 2: Workflow-1\n\n(5 comments)\n\ni can link to both sources if you like but \nthe reason there is a delta is in general glace metadef are not keep in sync as we have no job to enforce it.\n\nill rework this and split out the video models that are not supported yet into a seperate patch","accounts_in_message":[],"_revision_number":2},{"id":"1326d8061efb9d8ec34cd256b82567a51509c385","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-01 22:21:49.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nOkay, so maybe I didn\u0027t read the spec closely enough, but it seems like we need to determine what the source of truth for the values should be.\n\nIt actually kind of seems like we\u0027ve got two and are establishing a third, which kind of sucks.\n\nCan you give me a sense of the workflows that will be using these traits? Where will $entity_that_puts_them_on_resource_providers be looking to figure out which ones to put on resource providers?","accounts_in_message":[],"_revision_number":2},{"id":"e8152139266d56af947201a2229a1471766daa50","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-17 23:51:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ebfc7836b00aabb34a1f73ddb58ee6357311e93c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:11:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/96/652996/3/check/openstack-tox-pep8/f56f799/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/96/652996/3/check/openstack-tox-py27/261faad/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/96/652996/3/check/openstack-tox-docs/a7835aa/html/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/96/652996/3/check/openstack-tox-py36/164ca30/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/96/652996/3/check/openstack-tox-py37/2a10908/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"79ea2fd85c8bac121794e2c7edd18b8caf24df09","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-18 22:21:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\ntypo","accounts_in_message":[],"_revision_number":3},{"id":"1e992a174241e436c77d0b51c2ac94b2c859a737","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-19 16:38:38.000000000","message":"Patch Set 3: Workflow-1\n\n(2 comments)\n\nill update this shortly","accounts_in_message":[],"_revision_number":3},{"id":"6a77bae7380531b92d1068e34e7d4ec311f73c17","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-20 01:07:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"83704828db8bfc7276e1b506f4ee9069ef22cfdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 01:18:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/96/652996/4/check/openstack-tox-pep8/3af1c87/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/96/652996/4/check/openstack-tox-py27/e37fdb4/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/96/652996/4/check/openstack-tox-docs/368a547/html/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/96/652996/4/check/openstack-tox-py36/04dcbe2/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/96/652996/4/check/openstack-tox-py37/5bea66f/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"35624df09df69b1f2d01e435608e475f66defe58","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-21 03:52:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"08ad11558d96dae3cd6ea62e69baecd82facc55f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-24 10:38:43.000000000","message":"Patch Set 4: Code-Review+2\n\nNot approving yet in case cdent wants to come back around to this.","accounts_in_message":[],"_revision_number":4},{"id":"376e6bbd3da7fbdec5d9dcc1b92d135ce26b23b4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-24 13:12:53.000000000","message":"Patch Set 4: Workflow+1\n\nChris is on vacation this week, and it looks like his concerns were addressed.","accounts_in_message":[],"_revision_number":4},{"id":"b58206b5a45b556ee36f8a0a697c1978b890e2b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 13:13:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0f0d6853ff4968e7df2842efb030401a3fb54fee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 13:25:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"35df8a51e6cce4f50d0fdaee8b7003293196a5aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 13:25:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/96/652996/4/gate/openstack-tox-pep8/81440bd/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/96/652996/4/gate/openstack-tox-py27/4220311/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/96/652996/4/gate/openstack-tox-docs/0b44fbf/html/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/96/652996/4/gate/openstack-tox-py36/fe87575/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/96/652996/4/gate/openstack-tox-py37/be9faa4/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"21f232bd1c78459fc79c19fd2b4f2defdb758a63","revisions":{"d483b89071741c5b44d1740e252afafe27e7b02a":{"kind":"REWORK","_number":1,"created":"2019-04-16 13:43:38.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/96/652996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/96/652996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/96/652996/1"}}},"commit":{"parents":[{"commit":"1b14058122aa62439fb23de6efb10cce4a3f0c84","subject":"Add contextual comment about image type traits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/1b14058122aa62439fb23de6efb10cce4a3f0c84"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-04-16 12:45:44.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-04-16 13:43:35.000000000","tz":60},"subject":"add libvirt image metadata traits","message":"add libvirt image metadata traits\n\n- This change introduces standard traits for the\n  well defiend hw_disk_model, hw_cdrom_model, hw_video_model\n  and hw_vif_model image metadata properties.\n- This spec defines all of the new traits in the compute\n  namespace as they represent the ablity of a hypervisor to\n  emulate a specifc virtual device model, not the presence\n  of the respectiv hardware device on the host.\n\nChange-Id: I42237b196644bfa87701f978a1519941d447b450\nTask: 30526\nStory: 2005447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/d483b89071741c5b44d1740e252afafe27e7b02a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/d483b89071741c5b44d1740e252afafe27e7b02a"}]},"branch":"refs/heads/master"},"b22117170bbf347d2f3b0030436f5e3910506fc2":{"kind":"REWORK","_number":2,"created":"2019-05-01 19:40:08.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/96/652996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/96/652996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/96/652996/2"}}},"commit":{"parents":[{"commit":"1b14058122aa62439fb23de6efb10cce4a3f0c84","subject":"Add contextual comment about image type traits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/1b14058122aa62439fb23de6efb10cce4a3f0c84"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-04-16 12:45:44.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-05-01 19:40:08.000000000","tz":60},"subject":"add libvirt image metadata traits","message":"add libvirt image metadata traits\n\n- This change introduces standard traits for the\n  well defiend hw_disk_model, hw_cdrom_model, hw_video_model\n  and hw_vif_model image metadata properties.\n- This spec defines all of the new traits in the compute\n  namespace as they represent the ablity of a hypervisor to\n  emulate a specifc virtual device model, not the presence\n  of the respectiv hardware device on the host.\n\nChange-Id: I42237b196644bfa87701f978a1519941d447b450\nTask: 30526\nStory: 2005447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/b22117170bbf347d2f3b0030436f5e3910506fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/b22117170bbf347d2f3b0030436f5e3910506fc2"}]},"branch":"refs/heads/master"},"cda9e13f6efdd38a20f047634a20b008509c0178":{"kind":"REWORK","_number":3,"created":"2019-06-17 23:51:28.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/96/652996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/96/652996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/96/652996/3"}}},"commit":{"parents":[{"commit":"af871d1ed5f41ad0bf61a0ae6a8e639af8fcaf32","subject":"Create trait for NUMA subtree affinity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/af871d1ed5f41ad0bf61a0ae6a8e639af8fcaf32"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-04-16 12:45:44.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-06-17 23:51:16.000000000","tz":60},"subject":"add libvirt image metadata traits","message":"add libvirt image metadata traits\n\n- This change introduces standard traits for the\n  well defined hw_disk_model, hw_cdrom_model, hw_video_model\n  and hw_vif_model image metadata properties.\n- This spec defines all of the new traits in the compute\n  namespace as they represent the ability of a hypervisor to\n  emulate a specific virtual device model, not the presence\n  of the respective hardware device on the host.\n\nChange-Id: I42237b196644bfa87701f978a1519941d447b450\nTask: 30526\nStory: 2005447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/cda9e13f6efdd38a20f047634a20b008509c0178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/cda9e13f6efdd38a20f047634a20b008509c0178"}]},"branch":"refs/heads/master"},"21f232bd1c78459fc79c19fd2b4f2defdb758a63":{"kind":"REWORK","_number":4,"created":"2019-06-20 01:07:16.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/96/652996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/96/652996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/96/652996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/96/652996/4"}}},"commit":{"parents":[{"commit":"af871d1ed5f41ad0bf61a0ae6a8e639af8fcaf32","subject":"Create trait for NUMA subtree affinity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/af871d1ed5f41ad0bf61a0ae6a8e639af8fcaf32"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-04-16 12:45:44.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-06-20 01:07:05.000000000","tz":60},"subject":"add libvirt image metadata traits","message":"add libvirt image metadata traits\n\n- This change introduces standard traits for the\n  well defined hw_disk_model, hw_cdrom_model, hw_video_model\n  and hw_vif_model image metadata properties.\n- This spec defines all of the new traits in the compute\n  namespace as they represent the ability of a hypervisor to\n  emulate a specific virtual device model, not the presence\n  of the respective hardware device on the host.\n\nChange-Id: I42237b196644bfa87701f978a1519941d447b450\nTask: 30526\nStory: 2005447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/21f232bd1c78459fc79c19fd2b4f2defdb758a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/21f232bd1c78459fc79c19fd2b4f2defdb758a63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
