)]}'
{"id":"openstack%2Fos-traits~655671","triplet_id":"openstack%2Fos-traits~master~I6b652c20ba4f5ec775829a45939d708066dc3011","project":"openstack/os-traits","branch":"master","topic":"bp/amd-sev-libvirt-support","hashtags":[],"change_id":"I6b652c20ba4f5ec775829a45939d708066dc3011","subject":"Update SEV trait docs to avoid misleading people","status":"MERGED","created":"2019-04-25 11:12:37.000000000","updated":"2019-05-10 18:22:10.000000000","submitted":"2019-05-10 18:22:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655671-1557512530622-7f1a7d1e","meta_rev_id":"57d132ca2164fac998b0461a3ca50a1b7b59da52","_number":655671,"virtual_id_number":655671,"owner":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"date":"2019-04-25 17:21:29.000000000","_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-05-10 18:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2019-04-26 01:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":2,"date":"2019-04-25 17:31:27.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},{"value":2,"date":"2019-05-10 18:00:41.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-26 14:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},{"value":1,"date":"2019-05-10 18:00:41.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 11:13:14.000000000","updated_by":{"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},"reviewer":{"_account_id":24938,"name":"Bryan Stephenson","email":"bryan.stephenson@suse.com","username":"bryans"},"state":"REVIEWER"},{"updated":"2019-04-25 11:13:20.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"REVIEWER"},{"updated":"2019-04-25 11:13:41.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2019-04-25 12:05:39.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-04-25 17:31:27.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-26 01:49:02.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-04-26 14:36:46.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-05-10 18:00:41.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-05-10 18:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed28c67e302105e6f4178139c30efa7de0ecc0c0","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 11:12:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14e4b96812b9ef0c1caf0f8bf63eb4455ed5673e","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 11:15:31.000000000","message":"Patch Set 1:\n\nThis is a step in the right direction.  In https://review.opendev.org/#/c/655193/ there is talk of changing the namespace hierarchy which I don\u0027t think should block this review; however it might well require further tweaks or fixing merge conflicts depending on which lands first.","accounts_in_message":[],"_revision_number":1},{"id":"169057d1e80d049815874ac3c1070a84903ca1c1","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 11:20:40.000000000","message":"Patch Set 1:\n\nSee also https://review.opendev.org/#/c/655673/","accounts_in_message":[],"_revision_number":1},{"id":"0a4470639a6b9693bfddfd230dfe23002ede7e3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:23:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/71/655671/1/check/openstack-tox-pep8/20e2580/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/71/655671/1/check/openstack-tox-docs/c8e6609/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4706ef8b994552947619f9fccab38c3f7d0d231c","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-25 12:05:39.000000000","message":"Patch Set 1: Code-Review+2\n\n++","accounts_in_message":[],"_revision_number":1},{"id":"da6755cca13a3603b0532f461b8011d86103d204","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 14:44:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"177877ca366343f9c92d92d6257a319636230c23","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 15:54:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ff0afe3871317ac49f5900e02c58c31094715f4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 16:04:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d16ef6b49fa59507af036495679673f329c9510f","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 17:21:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1522205f88609dbfbc052beb26998043433f1425","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 17:21:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa760867a3798379bef6de8e175594f95c972fe5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 17:31:27.000000000","message":"Patch Set 2: Code-Review+2\n\nWorks.\n\nDon\u0027t forget to fix this when we deprecate it and introduce the new path.","accounts_in_message":[],"_revision_number":2},{"id":"c36710ff12121446412d561c6e1c0fccbf4c9c14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 17:33:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/71/655671/2/check/openstack-tox-pep8/dc3035a/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/71/655671/2/check/openstack-tox-docs/bad2278/html/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"df733b8eedb76e7d5024f64b204d67e89b47d6f8","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-04-26 01:49:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"109d98cd4e50e5fde7787caf53cd372072213ab8","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-04-26 14:36:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8d81a8e229d654bfda83958703100c77f59bf92c","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-10 18:00:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nfair enough","accounts_in_message":[],"_revision_number":2},{"id":"589ca3f0207e6c7dbb7c4be6b863690e6a69bd7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:00:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"921a1872c0d13e258207cdf3cf898151cccfeb3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:22:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/71/655671/2/gate/openstack-tox-pep8/b37ed50/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/71/655671/2/gate/openstack-tox-docs/132826b/html/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8a83283cc4163857cd2bdff773e7f4ae05d556c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:22:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3b9116739d5ed60760c5bbf59cceb845449eb931","revisions":{"52c92ddd939bb31cc716af1d21efde73e3a1affc":{"kind":"REWORK","_number":1,"created":"2019-04-25 11:12:37.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/71/655671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/71/655671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/71/655671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/71/655671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/71/655671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/71/655671/1"}}},"commit":{"parents":[{"commit":"50ca62cfbd635955f25f79004079ea9faba8edb1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/50ca62cfbd635955f25f79004079ea9faba8edb1"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 10:42:15.000000000","tz":60},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 10:42:15.000000000","tz":60},"subject":"Update SEV trait docs to avoid misleading people","message":"Update SEV trait docs to avoid misleading people\n\nSince the AMD SEV spec was approved for Stein, it was subsequently\nrealised that a trait would not be sufficient for tracking SEV-capable\ncompute hosts.  A resource class is also needed to track the inventory\nof \"slots\" available on these hosts, since the number of slots limits\nhow many guests with encrypted memory can run concurrently.\n\nTherefore the design pivoted somewhat, and now\ntrait:HW_CPU_AMD_SEV\u003drequired will not be the correct way to request\nSEV functionality:\n\n  https://specs.openstack.org/openstack/nova-specs/specs/train/approved/amd-sev-libvirt-support.html\n\nFor reference, the previous spec is here:\n\n  https://specs.openstack.org/openstack/nova-specs/specs/stein/approved/amd-sev-libvirt-support.html\n\nAnother lesson learnt from the Stein cycle was that it is not safe to\nassume that the work targeted for one cycle will actually land in that\ncycle, therefore it\u0027s safer for documentation of an in-progress\nfeature to be transparent that it\u0027s in progress and not yet usable.\n\nChange-Id: I6b652c20ba4f5ec775829a45939d708066dc3011\nblueprint: amd-sev-libvirt-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/52c92ddd939bb31cc716af1d21efde73e3a1affc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/52c92ddd939bb31cc716af1d21efde73e3a1affc"}]},"branch":"refs/heads/master"},"3b9116739d5ed60760c5bbf59cceb845449eb931":{"kind":"REWORK","_number":2,"created":"2019-04-25 17:21:04.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/71/655671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/71/655671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/71/655671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/71/655671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/71/655671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/71/655671/2"}}},"commit":{"parents":[{"commit":"50ca62cfbd635955f25f79004079ea9faba8edb1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/50ca62cfbd635955f25f79004079ea9faba8edb1"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 10:42:15.000000000","tz":60},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 17:19:02.000000000","tz":60},"subject":"Update SEV trait docs to avoid misleading people","message":"Update SEV trait docs to avoid misleading people\n\nSince the AMD SEV spec was approved for Stein, it was subsequently\nrealised that a trait would not be sufficient for tracking SEV-capable\ncompute hosts.  A resource class is also needed to track the inventory\nof \"slots\" available on these hosts, since the number of slots limits\nhow many guests with encrypted memory can run concurrently.\n\nTherefore the design pivoted somewhat, and now\ntrait:HW_CPU_AMD_SEV\u003drequired will not be the correct way to request\nSEV functionality:\n\n  https://specs.openstack.org/openstack/nova-specs/specs/train/approved/amd-sev-libvirt-support.html\n\nFor reference, the previous spec is here:\n\n  https://specs.openstack.org/openstack/nova-specs/specs/stein/approved/amd-sev-libvirt-support.html\n\nAnother lesson learnt from the Stein cycle was that it is not safe to\nassume that the work targeted for one cycle will actually land in that\ncycle, therefore it\u0027s safer for documentation of an in-progress\nfeature to be transparent that it\u0027s in progress and not yet usable.\n\nChange-Id: I6b652c20ba4f5ec775829a45939d708066dc3011\nblueprint: amd-sev-libvirt-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/3b9116739d5ed60760c5bbf59cceb845449eb931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/3b9116739d5ed60760c5bbf59cceb845449eb931"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
