)]}'
{"id":"openstack%2Fos-traits~655673","triplet_id":"openstack%2Fos-traits~master~I4028c25ff5fc0adeca6ee62b0db11d8bbf909f28","project":"openstack/os-traits","branch":"master","topic":"doc-no-trait-removal","hashtags":[],"change_id":"I4028c25ff5fc0adeca6ee62b0db11d8bbf909f28","subject":"Document policy of never removing traits","status":"MERGED","created":"2019-04-25 11:19:32.000000000","updated":"2019-04-25 14:56:04.000000000","submitted":"2019-04-25 13:48:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655673-1556200095435-eee29c9a","meta_rev_id":"c907ff1eddf2dc99a69243ed24cff31865610cb5","_number":655673,"virtual_id_number":655673,"owner":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-04-25 13:48:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-04-25 13:28:45.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-04-25 13:28:45.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 11:20:54.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-04-25 11:22:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-25 12:08:15.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-04-25 12:28:12.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-04-25 13:28:45.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-25 13:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2f0e16409f925c98e4d781ba6e1efffd9c76645","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 11:19:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70405795624829fad61e2b3d5066b75053485f1b","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 11:23:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e87724ae4ebe3a4e492431cdbed51ddb5e61bfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:32:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/73/655673/2/check/openstack-tox-pep8/01d728d/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/73/655673/2/check/openstack-tox-docs/b3df3b3/html/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a3c8534dbacc6a315c5d849ab59f7f540aa67a75","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-04-25 12:08:15.000000000","message":"Patch Set 2: Code-Review+2\n\nmuy bueno.","accounts_in_message":[],"_revision_number":2},{"id":"a640f6566de8021375484b6ef0f5e3f6a9d26503","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-04-25 12:28:12.000000000","message":"Patch Set 2: Code-Review+1\n\nyep good to get this tribal knowledge into the docs.","accounts_in_message":[],"_revision_number":2},{"id":"f47e701ca33cdf14ccd5da1b2a49953ecd3ac17e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 12:28:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nOne typo, but yes, this is good stuff.","accounts_in_message":[],"_revision_number":2},{"id":"e5d6614f4430a4d69c8d9ade4e1bb5b38c92cefe","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 13:28:25.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7c3378659ba713bf3dfffc6a9da7f812ba8ccd29","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-25 13:28:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nFixed typo, Jay and Sean were on board, fast approving.","accounts_in_message":[],"_revision_number":3},{"id":"70bc53777a73e68a96da6ea1a34a18f34ace9e57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 13:37:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/73/655673/3/check/openstack-tox-pep8/d046f03/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/73/655673/3/check/openstack-tox-docs/ad53194/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"036520af0cc31a15118b46777eb3cd70087f223f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 13:37:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1568a251a85264411d68a128bd1f821f1833a246","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 13:48:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3dae5fe033900663c682dc65152d46588d413192","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 13:48:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/73/655673/3/gate/openstack-tox-pep8/2881581/ : SUCCESS in 7m 50s\n- openstack-tox-docs http://logs.openstack.org/73/655673/3/gate/openstack-tox-docs/f6bd4d1/html/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"173bfbd91c860035c809fc982b3b6e28fbcf280f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-25 13:59:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"02faa6d30ad9de283a88df6d0166c3a742df7143","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-25 14:56:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f17a1ce37cce4f0950c330820a41077456a0f897","revisions":{"7615f00f8d52659c5157bbfe6c5dc42b74cec5b1":{"kind":"REWORK","_number":1,"created":"2019-04-25 11:19:32.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/73/655673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/73/655673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/73/655673/1"}}},"commit":{"parents":[{"commit":"50ca62cfbd635955f25f79004079ea9faba8edb1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/50ca62cfbd635955f25f79004079ea9faba8edb1"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 11:16:13.000000000","tz":60},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 11:16:13.000000000","tz":60},"subject":"Document policy of never removing traits","message":"Document policy of never removing traits\n\nIt has been asked at least twice in recent weeks (the first time by\nme) why unused traits can\u0027t be removed from os-traits:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-04-18.log.html#t2019-04-18T17:59:58\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-04-25.log.html#t2019-04-25T10:48:18\n\nso let\u0027s explicitly document this policy to prevent future generations\ngetting confused, or worse, attempting to remove traits.\n\nChange-Id: I4028c25ff5fc0adeca6ee62b0db11d8bbf909f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/7615f00f8d52659c5157bbfe6c5dc42b74cec5b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/7615f00f8d52659c5157bbfe6c5dc42b74cec5b1"}]},"branch":"refs/heads/master"},"b328885f9879b739028ace69d11d090135abb952":{"kind":"REWORK","_number":2,"created":"2019-04-25 11:23:27.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/73/655673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/73/655673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/73/655673/2"}}},"commit":{"parents":[{"commit":"50ca62cfbd635955f25f79004079ea9faba8edb1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/50ca62cfbd635955f25f79004079ea9faba8edb1"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 11:16:13.000000000","tz":60},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 11:23:11.000000000","tz":60},"subject":"Document policy of never removing traits","message":"Document policy of never removing traits\n\nIt has been asked at least twice in recent weeks (the first time by\nme) why unused traits can\u0027t be removed from os-traits:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-04-18.log.html#t2019-04-18T17:59:58\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-04-25.log.html#t2019-04-25T10:48:18\n\nso let\u0027s explicitly document this policy to prevent future generations\ngetting confused, or worse, attempting to remove traits.\n\nChange-Id: I4028c25ff5fc0adeca6ee62b0db11d8bbf909f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/b328885f9879b739028ace69d11d090135abb952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/b328885f9879b739028ace69d11d090135abb952"}]},"branch":"refs/heads/master"},"f17a1ce37cce4f0950c330820a41077456a0f897":{"kind":"REWORK","_number":3,"created":"2019-04-25 13:28:25.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/73/655673/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/73/655673/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/655673/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/73/655673/3"}}},"commit":{"parents":[{"commit":"50ca62cfbd635955f25f79004079ea9faba8edb1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/50ca62cfbd635955f25f79004079ea9faba8edb1"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-25 11:16:13.000000000","tz":60},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-04-25 13:28:23.000000000","tz":0},"subject":"Document policy of never removing traits","message":"Document policy of never removing traits\n\nIt has been asked at least twice in recent weeks (the first time by\nme) why unused traits can\u0027t be removed from os-traits:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-04-18.log.html#t2019-04-18T17:59:58\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-04-25.log.html#t2019-04-25T10:48:18\n\nso let\u0027s explicitly document this policy to prevent future generations\ngetting confused, or worse, attempting to remove traits.\n\nChange-Id: I4028c25ff5fc0adeca6ee62b0db11d8bbf909f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/f17a1ce37cce4f0950c330820a41077456a0f897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/f17a1ce37cce4f0950c330820a41077456a0f897"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
