)]}'
{"id":"openstack%2Fos-traits~656276","triplet_id":"openstack%2Fos-traits~master~I03a4c0fea621efa5efdd9bb8bfef3acfcfaee6fb","project":"openstack/os-traits","branch":"master","topic":"CPU_traits_to_mitigate_from_Meltdown_Spectre","hashtags":[],"change_id":"I03a4c0fea621efa5efdd9bb8bfef3acfcfaee6fb","subject":"ADD \"SVM\" CPU trait to amd.py","status":"ABANDONED","created":"2019-04-29 09:35:31.000000000","updated":"2019-05-27 15:46:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2b2933808497e572a9aaab68cf6e31902e698ac2","_number":656276,"virtual_id_number":656276,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":1,"date":"2019-04-29 15:05:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 09:36:44.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-04-29 09:36:51.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2019-04-29 09:36:57.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-29 09:37:09.000000000","updated_by":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"reviewer":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"state":"REVIEWER"},{"updated":"2019-04-29 15:05:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc06096295ccad214afbf7e515a17bd1a9faf6fc","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-29 09:35:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c91a2a8dad42b2185872b5e6b7edd8c635d03c09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 09:45:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/76/656276/1/check/openstack-tox-pep8/a729d0a/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/76/656276/1/check/openstack-tox-py27/fb3a734/ : FAILURE in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/76/656276/1/check/openstack-tox-docs/df360df/html/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/76/656276/1/check/openstack-tox-py36/d7a47b0/ : FAILURE in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/76/656276/1/check/openstack-tox-py37/1b4382d/ : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a555dec4d4f3cb2701857d4f812f3a86c65717af","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-29 14:49:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cab3eb927650b61758f092be663b20c5e0c771a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 15:05:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/76/656276/2/check/openstack-tox-pep8/ab4048f/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/76/656276/2/check/openstack-tox-py27/50fdbbc/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/76/656276/2/check/openstack-tox-docs/5126872/html/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/76/656276/2/check/openstack-tox-py36/658261f/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/76/656276/2/check/openstack-tox-py37/0092f57/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7fee235ef40e5d694936d0d0cc7912d9843539b5","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-27 15:46:16.000000000","message":"Abandoned\n\nIn favor of: \n\nhttps://review.opendev.org/655193","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7713c463a744752961607830a5d18fa56b217de","revisions":{"1d575850fbdcae7046774318589cff0081fdee33":{"kind":"REWORK","_number":1,"created":"2019-04-29 09:35:31.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/76/656276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/76/656276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/76/656276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/76/656276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/76/656276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/76/656276/1"}}},"commit":{"parents":[{"commit":"8127780c0d261e4afa965f226ca3a5c5f32ebb5f","subject":"Add CPU traits for Meltdown/Spectre mitigation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/8127780c0d261e4afa965f226ca3a5c5f32ebb5f"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-29 09:09:22.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-29 09:27:49.000000000","tz":120},"subject":"[Request for Comments] ADD \"SVM\" CPU trait to amd.py","message":"[Request for Comments] ADD \"SVM\" CPU trait to amd.py\n\nContext: We currently have two modules related to CPU traits:\n\n - amd.py: This has a single trait, \"SEV\", which is the AMD-specific CPU\n   flag.\n\n - x86.py: This file can \"imply\" that it is Intel-only (otherwise, why\n   would you have an \"amd.py\" file?  I realize that although the term\n   \"x86\" is Intel-based, it can also apply to AMD and other vendors who\n   hold an x86 architecture license).  So, as expected, the \"x86.py\"\n   file has mostly Intel-specific CPU traits, but ... unexpectedly, it\n   also has _one_ AMD-specific trait: \"SVM\" (Secure Virtual Machine).\n\nThe \u0027os_traits\u0027 water is now muddy.\n\nSo, as it stands, the \"SVM\" trait\u0027s module path is:\n\n    os_traits.HW_CPU_X86_SVM\n\nBut if we _move_ it to x86.py, its module path becomes:\n\n    os_traits.HW_CPU_AMD_SVM\n\nHowever, we can\u0027t do that -- because as per `os-traits` \"lifecycle\npolicy\"[1] `os_traits.HW_CPU_X86_SVM` cannot ever go away, because it\nwill cause breakage.\n\nGiven all this, I _think_ we want to consolidate all AMD-related traits\ninto a single file, amd.py, and all Intel-related ones into x86.py.  (And\nif a CPU trait is common between Intel and AMD, then they need to be\nadded to both amd.py and x86.py -- yes, this is repetitive, but maybe we\ncan live with it, unless we try Eric Fried\u0027s suggestion[2].\n\nFor now, add a note in x86.py saying that \"SVM\" is erroneously sitting\nthere.  And create a new trait in amd.py for \"SVM\".\n\n            * * *\n\n[1] https://docs.openstack.org/os-traits/latest/contributor/index.html\n\n[2] To try _three_ modules:\n\n      hw/cpu/x86.py (common to both Intel and AMD)\n      hw/cpu/x86/amd.py (AMD-specific)\n      hw/cpu/x86/intel.py (Intel-specific)\n\n    And Jay Pipes seem to prefer \"simplicity\" and go with just \"two\n    files\" approach.  If there\u0027s consensus on an approach brings\n    clarity, consistency and \"future-proofness\", I\u0027m all up for it.\n\nChange-Id: I03a4c0fea621efa5efdd9bb8bfef3acfcfaee6fb\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/1d575850fbdcae7046774318589cff0081fdee33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/1d575850fbdcae7046774318589cff0081fdee33"}]},"branch":"refs/heads/master"},"c7713c463a744752961607830a5d18fa56b217de":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2019-04-29 14:49:14.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/76/656276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/76/656276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/76/656276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/76/656276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/76/656276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/76/656276/2"}}},"commit":{"parents":[{"commit":"0c43e3005986d340dcaad660180a6117f7e2a468","subject":"Add CPU traits for Meltdown/Spectre mitigation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/0c43e3005986d340dcaad660180a6117f7e2a468"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-29 09:09:22.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-29 14:47:58.000000000","tz":120},"subject":"ADD \"SVM\" CPU trait to amd.py","message":"ADD \"SVM\" CPU trait to amd.py\n\nContext: We currently have two modules related to CPU traits:\n\n - amd.py: This has a single trait, \"SEV\", which is the AMD-specific CPU\n   flag.\n\n - x86.py: This file can \"imply\" that it is Intel-only (otherwise, why\n   would you have an \"amd.py\" file?  I realize that although the term\n   \"x86\" is Intel-based, it can also apply to AMD and other vendors who\n   hold an x86 architecture license).  So, as expected, the \"x86.py\"\n   file has mostly Intel-specific CPU traits, but ... unexpectedly, it\n   also has _one_ AMD-specific trait: \"SVM\" (Secure Virtual Machine).\n\nThe \u0027os_traits\u0027 water is now muddy.\n\nSo, as it stands, the \"SVM\" trait\u0027s module path is:\n\n    os_traits.HW_CPU_X86_SVM\n\nBut if we _move_ it to x86.py, its module path becomes:\n\n    os_traits.HW_CPU_AMD_SVM\n\nHowever, we can\u0027t do that -- because as per `os-traits` \"lifecycle\npolicy\"[1] `os_traits.HW_CPU_X86_SVM` cannot ever go away, because it\nwill cause breakage.\n\nGiven all this, I _think_ we want to consolidate all AMD-related traits\ninto a single file, amd.py, and all Intel-related ones into x86.py.  (And\nif a CPU trait is common between Intel and AMD, then they need to be\nadded to both amd.py and x86.py -- yes, this is repetitive, but maybe we\ncan live with it, unless we try Eric Fried\u0027s suggestion[2].\n\nFor now, add a note in x86.py saying that \"SVM\" is erroneously sitting\nthere.  And create a new trait in amd.py for \"SVM\".\n\n            * * *\n\n[1] https://docs.openstack.org/os-traits/latest/contributor/index.html\n\n[2] To try _three_ modules:\n\n      hw/cpu/x86.py (common to both Intel and AMD)\n      hw/cpu/x86/amd.py (AMD-specific)\n      hw/cpu/x86/intel.py (Intel-specific)\n\n    And Jay Pipes seem to prefer \"simplicity\" and go with just \"two\n    files\" approach.  If there\u0027s consensus on an approach brings\n    clarity, consistency and \"future-proofness\", I\u0027m all up for it.\n\nChange-Id: I03a4c0fea621efa5efdd9bb8bfef3acfcfaee6fb\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/c7713c463a744752961607830a5d18fa56b217de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/c7713c463a744752961607830a5d18fa56b217de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
