)]}'
{"id":"openstack%2Fos-traits~660168","triplet_id":"openstack%2Fos-traits~master~I127ed21d249675a1a876379bfcdb436749bf108b","project":"openstack/os-traits","branch":"master","topic":"Add_missing_CPU_traits","hashtags":[],"change_id":"I127ed21d249675a1a876379bfcdb436749bf108b","subject":"hw: cpu: Add certain missing traits","status":"ABANDONED","created":"2019-05-20 16:33:12.000000000","updated":"2019-05-27 15:45:17.000000000","total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2bac0d060b695d7259a8a9578e69e318f475762","_number":660168,"virtual_id_number":660168,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-21 08:47:39.000000000","_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":1,"date":"2019-05-20 18:12:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":-1,"date":"2019-05-20 21:23:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 16:35:10.000000000","updated_by":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"reviewer":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"state":"REVIEWER"},{"updated":"2019-05-20 18:12:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-20 21:23:28.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"}],"messages":[{"id":"7fdccbe706a6b57aa959d4b4239d5140f038cff3","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-20 16:33:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfa9a73c972f4816b9a7c0a895e220f1c55efd8b","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-20 16:34:27.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a2655576e3c34f4245e5baa7378a2a8ced22d416","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 16:44:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/660168/2/check/openstack-tox-pep8/09afa2c/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/68/660168/2/check/openstack-tox-py27/a35ebd6/ : FAILURE in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/68/660168/2/check/openstack-tox-docs/2e278bd/html/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/68/660168/2/check/openstack-tox-py36/670a672/ : FAILURE in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/68/660168/2/check/openstack-tox-py37/da490b8/ : FAILURE in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"31c207a65adcf42c40ae26b7ecd51ce71057d12e","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-20 18:02:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ed9578ffdfaf83bc1b671c8f3c28e8f6a0dd30af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 18:12:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/660168/3/check/openstack-tox-pep8/71a5a45/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/68/660168/3/check/openstack-tox-py27/fd77c76/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/68/660168/3/check/openstack-tox-docs/dee0da6/html/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/68/660168/3/check/openstack-tox-py36/21763bc/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/68/660168/3/check/openstack-tox-py37/4336f4f/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":3},{"id":"41c66f0ed5f1df7e4f54de924a6a77b1f060d6e4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-20 21:23:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(11 comments)\n\nNot sure why this isn\u0027t a new PS on [1]. Would be nice to have them consolidated so history is preserved.\n\n[1] https://review.opendev.org/#/c/655193/","accounts_in_message":[],"_revision_number":3},{"id":"3e8f8f6de78d8120539ec7838393f6e032877f0c","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-21 08:47:39.000000000","message":"Patch Set 3:\n\n(7 comments)\n\n(Can make this a new PS on https://review.opendev.org/#/c/655193/, to preserve history.)","accounts_in_message":[],"_revision_number":3},{"id":"fa6528d036e4755b14068dfd00db91e0ca21559b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-21 20:37:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7aaec4491c17deabd437009dcc3b9bdef7504b67","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-27 15:45:17.000000000","message":"Abandoned\n\nAs Eric correctly noted, I\u0027ve now made this change a new iteration of:\n\nhttps://review.opendev.org/655193\n\nAnd addressed the feedback from here.  So abandoning this.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9e185aefc08e191d74376d5ffebf737a57afb5e3","revisions":{"fdd439ad880ab33fcf8a52c6551bee12fc0f7097":{"kind":"REWORK","_number":1,"created":"2019-05-20 16:33:12.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/68/660168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/68/660168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/68/660168/1"}}},"commit":{"parents":[{"commit":"50ca62cfbd635955f25f79004079ea9faba8edb1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/50ca62cfbd635955f25f79004079ea9faba8edb1"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-20 15:33:38.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-20 16:28:21.000000000","tz":120},"subject":"hw: cpu: Add missing certain missing traits","message":"hw: cpu: Add missing certain missing traits\n\n[Work In Progress]\n                            ___......__             _\n                        _.-\u0027           ~-_       _.\u003da~~-_\n--\u003d\u003d\u003d\u003d\u003d-.-.-_----------~   .--.       _   -.__.-~ ( ___\u003d\u003d\u003e\n              \u0027\u0027\u0027--...__  (    \\ \\\\\\ { )       _.-~\n                        \u003d_ ~_  \\\\-~~~//~~~~-\u003d-~\n                         |-\u003d-~_ \\\\   \\\\\n                         |_/   \u003d. )   ~}\n                         |}      ||\n                        //       ||\n                      _//        {{\n                   \u0027\u003d\u0027~\u0027          \\\\_\n                                   ~~\u0027\n\n                        - - -\n\nThe main motivation for this change is to: (a) add missing CPU flags\n(including those flags that provide mitigation for the recent CVE flaws)\nas \u0027traits\u0027; and (b) adjust and clean up the layout of the \u0027hw/cpu/\u0027\ndirectory.\n\nTo that end, the following are the rough set of changes in this patch.\n(NB: Once we settle on the main content, I\u0027m amenable to split them as\nneeded.)\n\n(*) Introduce a new cpu/x86 directory; and vendor-specific files: amd.py\n    and intel.py; with __init__.py containing the *common* stuff (\n\n    - hw/cpu/x86/amd.py  (AMD-specific)\n    - hw/cpu/x86/intel.py  (Intel-specific)\n    - hw/cpu/x86/__init__.py  (Common for both AMD and Intel)\n\n       ---\n\n    - hw/cpy/x86.py  (Work out how to properly remove this file.\n      Contents of these are moved into x86/__init__.py.  Aside: I learnt\n      that it _is_ \"valid\" to have both x86.py, and x86/__init__.py; but\n      we should \"deprecate\" this file, too, for clarity\u0027s sake.)\n    - hw/cpu/amd.py  (Deprecate it with a comment.)\n\n(*) Add various missing CPU flags to x86/intel.py and x86/amd.py\n\n(*) Move, from x86.py, flags like \"VMX\" (Intel) and \"SVM\" (AMD) into\n    corresponding vendor-specific files.\n\nTODO: Comb-through all the flags again (especially the __init__.py) to\n      make sure you got only the common stuff; and have *not* regressed\n\nReferences:\n\n[1] Thread start:\n    http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006281.html\n    -- On reporting CPU flags that provide mitiation (to CVE flaws) as\n    Nova \u0027traits\u0027\n\n[2] Thread conclusion:\n    http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006364.html\n\nChange-Id: I127ed21d249675a1a876379bfcdb436749bf108b\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/fdd439ad880ab33fcf8a52c6551bee12fc0f7097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/fdd439ad880ab33fcf8a52c6551bee12fc0f7097"}]},"branch":"refs/heads/master"},"9759b9e9d53bc94e9513e0676557f99a23990cc8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-20 16:34:27.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/68/660168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/68/660168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/68/660168/2"}}},"commit":{"parents":[{"commit":"50ca62cfbd635955f25f79004079ea9faba8edb1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/50ca62cfbd635955f25f79004079ea9faba8edb1"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-20 15:33:38.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-20 16:34:15.000000000","tz":120},"subject":"hw: cpu: Add certain missing traits","message":"hw: cpu: Add certain missing traits\n\n[Work In Progress]\n                            ___......__             _\n                        _.-\u0027           ~-_       _.\u003da~~-_\n--\u003d\u003d\u003d\u003d\u003d-.-.-_----------~   .--.       _   -.__.-~ ( ___\u003d\u003d\u003e\n              \u0027\u0027\u0027--...__  (    \\ \\\\\\ { )       _.-~\n                        \u003d_ ~_  \\\\-~~~//~~~~-\u003d-~\n                         |-\u003d-~_ \\\\   \\\\\n                         |_/   \u003d. )   ~}\n                         |}      ||\n                        //       ||\n                      _//        {{\n                   \u0027\u003d\u0027~\u0027          \\\\_\n                                   ~~\u0027\n\n                        - - -\n\nThe main motivation for this change is to: (a) add missing CPU flags\n(including those flags that provide mitigation for the recent CVE flaws)\nas \u0027traits\u0027; and (b) adjust and clean up the layout of the \u0027hw/cpu/\u0027\ndirectory.\n\nTo that end, the following are the rough set of changes in this patch.\n(NB: Once we settle on the main content, I\u0027m amenable to split them as\nneeded.)\n\n(*) Introduce a new cpu/x86 directory; and vendor-specific files: amd.py\n    and intel.py; with __init__.py containing the *common* stuff (\n\n    - hw/cpu/x86/amd.py  (AMD-specific)\n    - hw/cpu/x86/intel.py  (Intel-specific)\n    - hw/cpu/x86/__init__.py  (Common for both AMD and Intel)\n\n       ---\n\n    - hw/cpy/x86.py  (Work out how to properly remove this file.\n      Contents of these are moved into x86/__init__.py.  Aside: I learnt\n      that it _is_ \"valid\" to have both x86.py, and x86/__init__.py; but\n      we should \"deprecate\" this file, too, for clarity\u0027s sake.)\n    - hw/cpu/amd.py  (Deprecate it with a comment.)\n\n(*) Add various missing CPU flags to x86/intel.py and x86/amd.py\n\n(*) Move, from x86.py, flags like \"VMX\" (Intel) and \"SVM\" (AMD) into\n    corresponding vendor-specific files.\n\nTODO: Comb-through all the flags again (especially the __init__.py) to\n      make sure you got only the common stuff; and have *not* regressed\n\nReferences:\n\n[1] Thread start:\n    http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006281.html\n    -- On reporting CPU flags that provide mitiation (to CVE flaws) as\n    Nova \u0027traits\u0027\n\n[2] Thread conclusion:\n    http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006364.html\n\nChange-Id: I127ed21d249675a1a876379bfcdb436749bf108b\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/9759b9e9d53bc94e9513e0676557f99a23990cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/9759b9e9d53bc94e9513e0676557f99a23990cc8"}]},"branch":"refs/heads/master"},"9e185aefc08e191d74376d5ffebf737a57afb5e3":{"kind":"REWORK","_number":3,"created":"2019-05-20 18:02:52.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/68/660168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/68/660168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/68/660168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/68/660168/3"}}},"commit":{"parents":[{"commit":"50ca62cfbd635955f25f79004079ea9faba8edb1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/50ca62cfbd635955f25f79004079ea9faba8edb1"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-20 15:33:38.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-20 18:01:44.000000000","tz":120},"subject":"hw: cpu: Add certain missing traits","message":"hw: cpu: Add certain missing traits\n\n[Work In Progress]\n                            ___......__             _\n                        _.-\u0027           ~-_       _.\u003da~~-_\n--\u003d\u003d\u003d\u003d\u003d-.-.-_----------~   .--.       _   -.__.-~ ( ___\u003d\u003d\u003e\n              \u0027\u0027\u0027--...__  (    \\ \\\\\\ { )       _.-~\n                        \u003d_ ~_  \\\\-~~~//~~~~-\u003d-~\n                         |-\u003d-~_ \\\\   \\\\\n                         |_/   \u003d. )   ~}\n                         |}      ||\n                        //       ||\n                      _//        {{\n                   \u0027\u003d\u0027~\u0027          \\\\_\n                                   ~~\u0027\n\n                        - - -\n\nThe main motivation for this change is to: (a) add missing CPU flags\n(including those flags that provide mitigation for the recent CVE flaws)\nas \u0027traits\u0027; and (b) adjust and clean up the layout of the \u0027hw/cpu/\u0027\ndirectory.\n\nTo that end, the following are the rough set of changes in this patch.\n(NB: Once we settle on the main content, I\u0027m amenable to split them as\nneeded.)\n\n(*) Introduce a new cpu/x86 directory; and vendor-specific files: amd.py\n    and intel.py; with __init__.py containing the *common* stuff (\n\n    - hw/cpu/x86/amd.py  (AMD-specific)\n    - hw/cpu/x86/intel.py  (Intel-specific)\n    - hw/cpu/x86/__init__.py  (Common for both AMD and Intel)\n\n       ---\n\n    - hw/cpy/x86.py  (Contents of this file are moved into\n      x86/__init__.py.)\n    - hw/cpu/amd.py  (Deprecate it with a comment.)\n\n(*) Add various missing CPU flags to x86/intel.py and x86/amd.py\n\n(*) Move, from x86.py, flags like \"VMX\" (Intel) and \"SVM\" (AMD) into\n    corresponding vendor-specific files.\n\nReferences:\n\n[1] Thread start:\n    http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006281.html\n    -- On reporting CPU flags that provide mitiation (to CVE flaws) as\n    Nova \u0027traits\u0027\n\n[2] Thread conclusion:\n    http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006364.html\n\nChange-Id: I127ed21d249675a1a876379bfcdb436749bf108b\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/9e185aefc08e191d74376d5ffebf737a57afb5e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/9e185aefc08e191d74376d5ffebf737a57afb5e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
