)]}'
{"id":"openstack%2Fos-traits~662811","triplet_id":"openstack%2Fos-traits~master~I055f193988addd6b88f83ff317ea121a7d81063b","project":"openstack/os-traits","branch":"master","topic":"story/2005575","hashtags":[],"change_id":"I055f193988addd6b88f83ff317ea121a7d81063b","subject":"Create trait for device subtree affinity","status":"ABANDONED","created":"2019-06-03 17:38:03.000000000","updated":"2022-01-25 16:48:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"39831768d4c1b13f8ae85d05d8bf6c99d53dc7bb","_number":662811,"virtual_id_number":662811,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-04 15:50:34.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2019-06-03 17:49:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":1,"date":"2020-03-19 08:48:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"date":"2019-06-05 23:47:24.000000000","_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 17:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-05 23:47:24.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-03-19 08:48:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"a7804acbf531cfb179add7cfc1b0742ccb0de480","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-03 17:38:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afa407335aa9c385f9bbe552d728786b888e1175","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 17:49:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/11/662811/1/check/openstack-tox-pep8/8e9cd7b/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/11/662811/1/check/openstack-tox-py27/53fecc8/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/11/662811/1/check/openstack-tox-docs/4defa57/html/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/11/662811/1/check/openstack-tox-py36/ffeda6c/ : SUCCESS in 4m 53s\n- openstack-tox-py37 http://logs.openstack.org/11/662811/1/check/openstack-tox-py37/bc6dc03/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3b7f08044f0e9bb5f9cae002507da0ccc54da845","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-04 08:07:40.000000000","message":"Patch Set 1: Code-Review-1\n\nI don\u0027t see direct use of this trait yet.","accounts_in_message":[],"_revision_number":1},{"id":"45023c5a3665feea540fb4a0370936f7e6137f1e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-04 15:50:34.000000000","message":"Patch Set 1:\n\n\u003e I don\u0027t see direct use of this trait yet.\n\nI think Cyborg has a use case, but will let Sundar weigh in.","accounts_in_message":[],"_revision_number":1},{"id":"dabd8c3f80ca6469c9afeb3f799b14812d90961e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-05 07:01:27.000000000","message":"Patch Set 1:\n\n\u003e \u003e I don\u0027t see direct use of this trait yet.\n \u003e \n \u003e I think Cyborg has a use case, but will let Sundar weigh in.\n\nSure, let\u0027s get that use case defined then I\u0027m happy to unblock this.","accounts_in_message":[],"_revision_number":1},{"id":"4c857f1a46d43ae86edf263ae0c974f4b8c6cb9d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-05 23:47:24.000000000","message":"Patch Set 1:\n\nDevice co-location is certainly an important use case for upcoming devices. Specifically, a single device has multiple components (ASICs, FPGAs) which have different resource classes and traits. The resources for a VM should come from a single PCI card, even when multiple such cards are present in the system. Modeling each component as a RP with differing traits and RCs is the only feasible way to go. \n\nThe alternative is to model the entire device as a single RP, with all RCs and traits mixed. But that won\u0027t work when there are overlapping traits.","accounts_in_message":[],"_revision_number":1},{"id":"e9a03bf2ee6465cba0971055db46ec4099d2ae8b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-03-19 08:48:59.000000000","message":"Patch Set 1: Code-Review+1\n\nI see the use case now. but I guess this trait is not needed for Ussuri yet.","accounts_in_message":[],"_revision_number":1},{"id":"39831768d4c1b13f8ae85d05d8bf6c99d53dc7bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-25 16:48:34.000000000","message":"Abandoned\n\nwe no longer have any work in flight that will consume this so for now ill abandon this for now to keep the review backlog clean. we can revive it if contributors pick back up this effort in the future.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23b9a7f2d7774f079b965a3d85afb2d43b72deb8","revisions":{"23b9a7f2d7774f079b965a3d85afb2d43b72deb8":{"kind":"REWORK","_number":1,"created":"2019-06-03 17:38:03.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/11/662811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/11/662811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/11/662811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/11/662811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/11/662811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/11/662811/1"}}},"commit":{"parents":[{"commit":"08ba939ba575f61591b0b71022fa8dec62c02535","subject":"Merge \"hw: cpu: Rework the directory layout; add missing traits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/08ba939ba575f61591b0b71022fa8dec62c02535"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-03 17:36:20.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-03 17:37:55.000000000","tz":-300},"subject":"Create trait for device subtree affinity","message":"Create trait for device subtree affinity\n\nIntroduce trait HW_DEVICE_ROOT. This is to be used with the\n\u0027resourceless RP\u0027 and \u0027same_subtree\u0027 placement features [1] when\nrequesting affinity (e.g. multiple VFs from PFs on the same NIC)\nespecially when not requesting resources from the subtree root provider\nitself. See for example [2], where HW_DEVICE_ROOT is glibly named\nI_AM_A_NIC.\n\n[1] https://review.opendev.org/#/c/662191/\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006059.html\n\nChange-Id: I055f193988addd6b88f83ff317ea121a7d81063b\nStory: #2005575\nTask: #30908\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/23b9a7f2d7774f079b965a3d85afb2d43b72deb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/23b9a7f2d7774f079b965a3d85afb2d43b72deb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
