)]}'
{"id":"openstack%2Fos-traits~673402","triplet_id":"openstack%2Fos-traits~master~I6afd7ef33edd5187c6242d0e0e58c871cb83250e","project":"openstack/os-traits","branch":"master","topic":"bp/expose-auto-converge-post-copy","hashtags":[],"change_id":"I6afd7ef33edd5187c6242d0e0e58c871cb83250e","subject":"Add new traits to flag live migration features","status":"MERGED","created":"2019-07-30 05:23:33.000000000","updated":"2019-07-30 13:37:27.000000000","submitted":"2019-07-30 13:37:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673402-1564493847168-fe339b59","meta_rev_id":"28dca0346068634143ebcdcdb6a5cb6d55094c1c","_number":673402,"virtual_id_number":673402,"owner":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-07-30 13:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":2,"date":"2019-07-30 13:20:09.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-07-30 05:49:56.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2019-07-30 13:22:11.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"date":"2019-07-30 07:51:52.000000000","_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":1,"date":"2019-07-30 13:22:11.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 05:49:56.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-07-30 07:51:52.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2019-07-30 13:20:09.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-30 13:22:11.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-07-30 13:37:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e822fab197df2d7a624fdffeedfc4b822f6350e","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-07-30 05:23:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12a0f1c003518d6da18f18241b91901eac21e98e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 05:31:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/02/673402/1/check/openstack-tox-pep8/039761c/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/02/673402/1/check/openstack-tox-py27/c634f90/ : SUCCESS in 2m 38s\n- openstack-tox-docs http://logs.openstack.org/02/673402/1/check/openstack-tox-docs/429e71a/html/ : SUCCESS in 2m 02s\n- openstack-tox-py36 http://logs.openstack.org/02/673402/1/check/openstack-tox-py36/bfb1c5c/ : SUCCESS in 2m 32s\n- openstack-tox-py37 http://logs.openstack.org/02/673402/1/check/openstack-tox-py37/93e7234/ : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"778b5e86af48fae6935dbb822926f710511e6999","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-07-30 05:38:11.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"89433e213370543e70a86d55d5fd21cb5da53d1c","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-07-30 05:39:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de50f8ba4da8bb754d6aa70970a04ca5bd3933c4","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2019-07-30 05:48:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9781260b258c513917c6a9e4ff249f21b3e23869","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-07-30 05:49:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7e9bf312b957d8083138a4c25be795c345f13f4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 05:52:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/02/673402/2/check/openstack-tox-pep8/ad5f4ba/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/02/673402/2/check/openstack-tox-py27/998a785/ : SUCCESS in 2m 48s\n- openstack-tox-docs http://logs.openstack.org/02/673402/2/check/openstack-tox-docs/a6b98e2/html/ : SUCCESS in 2m 38s\n- openstack-tox-py36 http://logs.openstack.org/02/673402/2/check/openstack-tox-py36/b0db2f8/ : SUCCESS in 2m 56s\n- openstack-tox-py37 http://logs.openstack.org/02/673402/2/check/openstack-tox-py37/75849c6/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8a31f942c50a4e7685ccf0cf6c1cd6f6711da27a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-30 13:20:09.000000000","message":"Patch Set 2: Code-Review+2\n\nThis conforms to the spec [1].\n\nNeed non-Intel +A.\n\n[1] http://specs.openstack.org/openstack/nova-specs/specs/train/approved/expose-auto-converge-post-copy.html#proposed-change","accounts_in_message":[],"_revision_number":2},{"id":"585f85a61d6d75f126714da6f453623488fa05d6","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-30 13:22:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nhappy to approve but it needs a followup","accounts_in_message":[],"_revision_number":2},{"id":"3a7ac64b8b8548a2d4d76fc14d0a364b2637056d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 13:22:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b470aaa09b7e2f5cce3d95182481ddb392227798","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 13:37:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/02/673402/2/gate/openstack-tox-pep8/bf4c036/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/02/673402/2/gate/openstack-tox-py27/1cc5763/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/02/673402/2/gate/openstack-tox-docs/106dfcb/html/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/02/673402/2/gate/openstack-tox-py36/240dd3f/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/02/673402/2/gate/openstack-tox-py37/91b73c0/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1953902fef23d5728df9911f089ec23d325b7116","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 13:37:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9de1fa7b3b5dd6a34a1823a780279c71b517feff","revisions":{"4072b0cbf8e3d64346e136aa95bf13c5b2c721dd":{"kind":"REWORK","_number":1,"created":"2019-07-30 05:23:33.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/02/673402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/02/673402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/02/673402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/02/673402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/02/673402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/02/673402/1"}}},"commit":{"parents":[{"commit":"3297a6692caccda239657080a6dc6579c0dce18b","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/3297a6692caccda239657080a6dc6579c0dce18b"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-07-24 15:20:19.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-07-30 05:23:08.000000000","tz":480},"subject":"Add new traits to flag live migration features","message":"Add new traits to flag live migration features\n\nAdd two new tratis:\n- COMPUTE_MIGRATE_AUTO_CONVERGE\n- COMPUTE_MIGRATE_POST_COPY\n\nChange-Id: I6afd7ef33edd5187c6242d0e0e58c871cb83250e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/4072b0cbf8e3d64346e136aa95bf13c5b2c721dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/4072b0cbf8e3d64346e136aa95bf13c5b2c721dd"}]},"branch":"refs/heads/master"},"9de1fa7b3b5dd6a34a1823a780279c71b517feff":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-30 05:48:41.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/02/673402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/02/673402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/02/673402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/02/673402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/02/673402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/02/673402/2"}}},"commit":{"parents":[{"commit":"3297a6692caccda239657080a6dc6579c0dce18b","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/3297a6692caccda239657080a6dc6579c0dce18b"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-07-24 15:20:19.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2019-07-30 05:48:13.000000000","tz":480},"subject":"Add new traits to flag live migration features","message":"Add new traits to flag live migration features\n\nAdd two new tratis:\n- COMPUTE_MIGRATE_AUTO_CONVERGE\n- COMPUTE_MIGRATE_POST_COPY\n\nImplements: blueprint expose-auto-converge-post-copy\n\nChange-Id: I6afd7ef33edd5187c6242d0e0e58c871cb83250e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/9de1fa7b3b5dd6a34a1823a780279c71b517feff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/9de1fa7b3b5dd6a34a1823a780279c71b517feff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
