)]}'
{"id":"openstack%2Fos-traits~686464","triplet_id":"openstack%2Fos-traits~master~I6309501f10544f468bb694361c845a393a9491cf","project":"openstack/os-traits","branch":"master","topic":"print_tree","hashtags":[],"change_id":"I6309501f10544f468bb694361c845a393a9491cf","subject":"Add method to visualize the os-traits tree","status":"MERGED","created":"2019-10-03 19:08:24.000000000","updated":"2019-10-09 15:57:17.000000000","submitted":"2019-10-09 15:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686464-1570636568744-70f2cce9","meta_rev_id":"ae5f8ac6e09649881fa200740a6cf8c69d47ff3e","_number":686464,"virtual_id_number":686464,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-10-08 18:33:54.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 15:56:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-10-09 08:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-09 15:43:52.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-09 15:43:52.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 08:25:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-09 15:43:52.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-10-09 15:56:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96d6e9ebe88096b692cb806ca56297ed08cdb344","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-03 19:08:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34e01a0f47183df7f425ac301e55f303e19c1292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:15:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4641a3e26a904cf08f9042755f661ac3 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d07560bdc0e4a6798d7f7fc37af3784 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde5e33188c14254aa4257bcd5743382 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2da27bc61e0b4933add0d3442aef95ed : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dbdf4ce79df44119638b8126216beed : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7893866e598a62331825055ce288347afa63bbcd","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-10-04 07:51:30.000000000","message":"Patch Set 1: Code-Review+2\n\nseems okay, doesn\u0027t seem to break anything, but i you wanted to be truly complete you\u0027d explain why you want such a thing in the commit message\n\nor at least tell me, because I\u0027m curious","accounts_in_message":[],"_revision_number":1},{"id":"0300d227db37ffd0cc17528eff7277089f64b9ee","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-04 13:09:03.000000000","message":"Patch Set 1:\n\n\u003e explain why you want such a thing\n\nYeah.\n\nI get questions (lately mainly from my downstreams) about traits, and one of them asked for an explanation of the taxonomy/nomenclature/hierarchy. I had thought there was already some easy way to visualize that, but found there wasn\u0027t. I started to roll a one-off and realized it would share a bunch of code we already had for generating the trait strings. One thing led to another, and nine months later...","accounts_in_message":[],"_revision_number":1},{"id":"c7a52e057761b4c36768cb19c1e33ab7da195fc9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-08 14:21:05.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nMain issue is whether we\u0027re breaking an API contract or not. Also not a huge fan of adding yet more stuff to the \u0027__init__\u0027 file, but we\u0027re already doing it and it might be hard to avoid given how we do things here so I can hold my nose on that one","accounts_in_message":[],"_revision_number":1},{"id":"02a7a0fcb1a3e08e718d49488198cdad1658d6ea","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 17:30:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d030ac3505e26bf071a235c20fc8e0a0a82439a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 17:35:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9fbda998037642c8ed60e48e78fd104f6dd4411","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 17:35:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04916c4904d9d5ed59af1f530ddf10dacd796c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 17:49:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e109a04ae78c4f9ebfe50f324361e41b : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bffe9077e0fa476593ffe7d1ede79c46 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9105bb20be544bc89480c3bd4591c43 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f675c6c482f44efb5004cefafbb1b04 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c0fd00367a54db28c47383e087d38f4 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"380b8a40b4016f89fbd16ee6284cc3985f835ddf","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 18:33:54.000000000","message":"Patch Set 2:\n\n\u003e make everything before L115 explicitly private\n\nhttps://review.opendev.org/#/c/687378/\n\n\u003e not in the docs anywhere\n\nhttps://storyboard.openstack.org/#!/story/2006682 (I didn\u0027t have it in me to actually go make this happen right away)","accounts_in_message":[],"_revision_number":2},{"id":"365886d9d81a7051a13321ecec1eb2d17f840f86","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-09 08:25:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a2032f30c014e012f71432671426f2c88896f69e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-10-09 15:43:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nsure","accounts_in_message":[],"_revision_number":2},{"id":"dfdfbcf997c255de827292bd70a008b89fcc8985","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:44:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a214451b7967243d06d36f45159d4087e3021ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:56:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"38fce0047646fd54f0da792f81b8a356b5032463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:56:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f769c6727cb444da8cc7ac9ac02ce180 : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02272e4eeae048628225b0fbf0812abb : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe62d68d9a44d04960c7e291bfdbced : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c99379aded4fd4909755c6ba71e985 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75545b1edec841bfad607576fc99fba6 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"e2653adf396cb6190ac0d8da8fb303c3684d27f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:57:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21440b1035f04403973d9f7fd25398a5 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a5e829b0d6579a0eed27df6f02a62f0ff371a8d","revisions":{"2136acae32c562b3f9d8f5a77a6db696400b7a04":{"kind":"REWORK","_number":1,"created":"2019-10-03 19:08:24.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/64/686464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/64/686464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/64/686464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/64/686464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/64/686464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/64/686464/1"}}},"commit":{"parents":[{"commit":"f47d88f085dcf3df49ff9e9d1609f00d75c24024","subject":"Merge \"Add support for ppc64le platforms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/f47d88f085dcf3df49ff9e9d1609f00d75c24024"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-03 19:04:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-03 19:04:15.000000000","tz":-300},"subject":"Add method to visualize the os-traits tree","message":"Add method to visualize the os-traits tree\n\nAdd a print_tree method in the base os_traits package. This recursively\nwalks the repository and prints a formatted tree of short names of all\nthe namespaces and their traits. You can try it like this:\n\n  python -c \u0027import os_traits; os_traits.print_tree()\u0027\n\nHere\u0027s an abbreviated subset of the output:\n\n   COMPUTE:\n      DEVICE_TAGGING, TRUSTED_CERTS\n      GRAPHICS:\n         MODEL_CIRRUS, MODEL_GOP, MODEL_NONE, MODEL_QXL, MODEL_VGA, MODEL_VIRTIO, MODEL_VMVGA, MODEL_XEN\n   HW:\n      CPU:\n         HYPERTHREADING\n         AMD:\n            SEV\n         PPC64LE:\n            POWER8, POWER9\n      NUMA:\n         ROOT\n   MISC:\n      SHARES_VIA_AGGREGATE\n   STORAGE:\n      DISK:\n         HDD, SSD\n\nChange-Id: I6309501f10544f468bb694361c845a393a9491cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/2136acae32c562b3f9d8f5a77a6db696400b7a04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/2136acae32c562b3f9d8f5a77a6db696400b7a04"}]},"branch":"refs/heads/master"},"6a5e829b0d6579a0eed27df6f02a62f0ff371a8d":{"kind":"REWORK","_number":2,"created":"2019-10-08 17:35:09.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/64/686464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/64/686464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/64/686464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/64/686464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/64/686464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/64/686464/2"}}},"commit":{"parents":[{"commit":"f47d88f085dcf3df49ff9e9d1609f00d75c24024","subject":"Merge \"Add support for ppc64le platforms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/f47d88f085dcf3df49ff9e9d1609f00d75c24024"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-03 19:04:15.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-08 17:35:06.000000000","tz":-300},"subject":"Add method to visualize the os-traits tree","message":"Add method to visualize the os-traits tree\n\nAdd a print_tree method in the base os_traits package. This recursively\nwalks the repository and prints a formatted tree of short names of all\nthe namespaces and their traits. You can try it like this:\n\n  python -c \u0027import os_traits; os_traits.print_tree()\u0027\n\nHere\u0027s an abbreviated subset of the output:\n\n   COMPUTE:\n      DEVICE_TAGGING, TRUSTED_CERTS\n      GRAPHICS:\n         MODEL_CIRRUS, MODEL_GOP, MODEL_NONE, MODEL_QXL, MODEL_VGA, MODEL_VIRTIO, MODEL_VMVGA, MODEL_XEN\n   HW:\n      CPU:\n         HYPERTHREADING\n         AMD:\n            SEV\n         PPC64LE:\n            POWER8, POWER9\n      NUMA:\n         ROOT\n   MISC:\n      SHARES_VIA_AGGREGATE\n   STORAGE:\n      DISK:\n         HDD, SSD\n\nChange-Id: I6309501f10544f468bb694361c845a393a9491cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/6a5e829b0d6579a0eed27df6f02a62f0ff371a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/6a5e829b0d6579a0eed27df6f02a62f0ff371a8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
