)]}'
{"id":"openstack%2Fos-traits~696435","triplet_id":"openstack%2Fos-traits~master~Ia9c7b83b7b163ce73e980923f48bc34045ff92be","project":"openstack/os-traits","branch":"master","topic":"bug_1854235","hashtags":[],"change_id":"Ia9c7b83b7b163ce73e980923f48bc34045ff92be","subject":"Add \u0027TYPE_PLOOP\u0027 image type.","status":"MERGED","created":"2019-11-28 02:10:48.000000000","updated":"2019-12-02 15:03:03.000000000","submitted":"2019-12-02 13:34:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696435-1575293665675-82587bb4","meta_rev_id":"96a7bcbd7fa92788afc28969e67022d1b698337e","_number":696435,"virtual_id_number":696435,"owner":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 13:34:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2019-12-02 13:15:45.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2019-12-02 13:27:14.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-29 01:40:52.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-11-29 15:26:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-02 13:27:14.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-12-02 13:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3b853952952a9e5b9657b780bba09b6924693f5","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2019-11-28 02:10:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61b182ca098e10b29113fa446a4d8899102955f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 02:19:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2daa4879f714b40887a37a7411ff003 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/561e4e18f52a49bf963170c9b8d8c33b : SUCCESS in 2m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66352cc6c3d24519bbc1cca1745ce151 : SUCCESS in 2m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b64376a0a02c429183b89ff3c0a9c196 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d56042d14930b1ea6d848eae9413b5595b598051","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-11-29 01:40:52.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c23fa61556a1b4a8c410a293ee7fa0562f45916a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-11-29 15:26:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nMakes sense to me","accounts_in_message":[],"_revision_number":1},{"id":"833449a46224524897a431222a291234b492f714","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-12-02 13:14:11.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI\u0027ll update the link and fast approve.","accounts_in_message":[],"_revision_number":1},{"id":"7a6b7e1d6b22c609dcccf6a44334936b9c876699","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-12-02 13:14:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"886197cbf3a526cfbc584e8869be0b57c2500e4d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-12-02 13:15:45.000000000","message":"Patch Set 2: Code-Review+2\n\nI just updated the link in the comment. Will give Stephen a couple minutes, otherwise will fast approve.","accounts_in_message":[],"_revision_number":2},{"id":"3298fd56a1ddc5f288b128006c94af59c6aa7a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:25:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f467ff1e9f3d4e868c0db6586bc88fd3 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b073be7b5948cdaedbc9587a6f527e : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/838a59f8aaa743d8b1d3c1e93263e9a7 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a395b4020b548949e1fc7a0912dffa0 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e1395304fc06d667bf5a37bf76b6bc39ebc1517f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-12-02 13:27:14.000000000","message":"Patch Set 2: Workflow+1\n\nrosmaita confirms the link is right. I\u0027m gonna send this.","accounts_in_message":[],"_revision_number":2},{"id":"b9e0ddaa61a40a40348b053b226b1737a9f1664c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:27:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad5436d13fbd9f647a2140ce6bead8bd25bb8b55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:34:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"691c94cce4d3394834c35282734383f69305a33e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:34:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/227b7f25168940b781e41f9a7786e518 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d0555b16b7d49f3aa22a2444283469d : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c8a938cee77410b8fb163fba9a7366f : SUCCESS in 3m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6357a91b30f40d0991d0aa062eeb598 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"59ab5d1d234f28ed68b4fc6565704fbe9162fa23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:35:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ddae605136b40ccac4dea2025208015 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2},{"id":"e92cdf0cc5d945a4b74c3383b91b53c231308ddd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-12-02 15:03:03.000000000","message":"Patch Set 2:\n\nI proposed a new release to pick this up https://review.opendev.org/#/c/696892/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7468b0bad407c7fdb293a49e9c6c9ec16274a96","revisions":{"e4bef367951f02a937b73ea4999dda47708bf9f8":{"kind":"REWORK","_number":1,"created":"2019-11-28 02:10:48.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/35/696435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/35/696435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/696435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/696435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/696435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/35/696435/1"}}},"commit":{"parents":[{"commit":"173298bf862ea13ed32d0642f235ecee23e5d3b0","subject":"Stop testing with py2 and switch to ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/173298bf862ea13ed32d0642f235ecee23e5d3b0"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2019-11-28 02:08:44.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2019-11-28 02:08:44.000000000","tz":0},"subject":"Add \u0027TYPE_PLOOP\u0027 image type.","message":"Add \u0027TYPE_PLOOP\u0027 image type.\n\nThis patch adds missing image type \u0027TYPE_PLOOP\u0027 for Virtuozzo.\n\nChange-Id: Ia9c7b83b7b163ce73e980923f48bc34045ff92be\nCloses-Bug: #1854235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/e4bef367951f02a937b73ea4999dda47708bf9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/e4bef367951f02a937b73ea4999dda47708bf9f8"}]},"branch":"refs/heads/master"},"c7468b0bad407c7fdb293a49e9c6c9ec16274a96":{"kind":"REWORK","_number":2,"created":"2019-12-02 13:14:37.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/35/696435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/35/696435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/696435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/696435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/696435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/35/696435/2"}}},"commit":{"parents":[{"commit":"173298bf862ea13ed32d0642f235ecee23e5d3b0","subject":"Stop testing with py2 and switch to ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/173298bf862ea13ed32d0642f235ecee23e5d3b0"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2019-11-28 02:08:44.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-12-02 13:14:33.000000000","tz":0},"subject":"Add \u0027TYPE_PLOOP\u0027 image type.","message":"Add \u0027TYPE_PLOOP\u0027 image type.\n\nThis patch adds missing image type \u0027TYPE_PLOOP\u0027 for Virtuozzo.\n\nChange-Id: Ia9c7b83b7b163ce73e980923f48bc34045ff92be\nCloses-Bug: #1854235\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/c7468b0bad407c7fdb293a49e9c6c9ec16274a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/c7468b0bad407c7fdb293a49e9c6c9ec16274a96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
