)]}'
{"id":"openstack%2Fos-traits~704535","triplet_id":"openstack%2Fos-traits~master~I337f9fefe7f0364bce3a0192dfc9c1199076fa02","project":"openstack/os-traits","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I337f9fefe7f0364bce3a0192dfc9c1199076fa02","subject":"setup.cfg: Remove unused \u0027build_sphinx\u0027 configuration","status":"MERGED","created":"2020-01-28 11:12:27.000000000","updated":"2020-02-27 16:38:04.000000000","submitted":"2020-02-27 16:34:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704535-1582821295875-97979107","meta_rev_id":"93297ceef359022fcf4af470c5eb964c630153a5","_number":704535,"virtual_id_number":704535,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 16:34:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2020-02-27 15:05:53.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2020-02-27 16:21:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2020-02-27 16:21:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 15:13:22.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-02-27 16:21:15.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-02-27 16:34:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"159d56fd05d8c7ec1ccf30fcc8518f58fc6bf647","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-28 11:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d047e76c4fd864a5810a04b3832252d1aa7bb495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 11:21:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e14f0e142f844be59677c9de7830ed64 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a7b55372b242c0a5af96830fe4ba24 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b10a3b44c154e7997c0bf07b83cde97 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae59279693244520ba8dacbc9300e931 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30b62999ebc24772b48ce5cf971eebcb : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97c637b3c6f748cfa08fe4a0dace9c33 : SUCCESS in 3m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3798ec697a9a42ade1d19433902402102fcf899a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-28 15:13:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de40fbbe8307375d544205a85462743e6fb8dbde","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-27 15:05:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7e5e1d761fc56a8e54cd8a30470913095cc469d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:54:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c530929cab194f90bbc35c56a9e504ab : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d471bff69594c649325e0b5079a70b8 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf82e49f8114d14a3eb7943b40dee19 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a7d1c09b707436e9c1e511de14808de : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b79ae8c1b57f4c74b4ba27621281dc5a : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5414934316045b19c8497ded3579edf : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3addc8d8605ffd6092816bdf4deb46a97c3e79db","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-27 16:21:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49376e382f9296a5bd0a1bbfe51a3211a9cb0db5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:21:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e34ccfa34e3b6991d777a6fcfcd42125c8894a8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:34:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/878035cc4b934340ad4529027382b14e : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc6e0b867514a01b3a48f941166fe5d : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3726f278096a47ebb0249604742ee0da : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/769b1198df334e92a4609ad7b5e7650d : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2655cbb91cba422bb734e29dd0e5587a : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"257edc4cad1b56df9689a3ee817785cfd517e303","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:34:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"54b4434f9464d2e1a1193e9de90a03adace30480","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:38:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d2a06a270a64091b771b188ee807f10 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4ee198eb7f95f7192d9d5dcb157289848bc2506","revisions":{"abba182e12889f162e2ad992aa66bd5dfb4a7fed":{"kind":"REWORK","_number":1,"created":"2020-01-28 11:12:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/704535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/35/704535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/704535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/704535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/704535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/35/704535/1"}}},"commit":{"parents":[{"commit":"2f4784d510a895b084421614b983333aeb63061d","subject":"setup.cfg: Configure \u0027python-requires\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/2f4784d510a895b084421614b983333aeb63061d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-28 11:10:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-28 11:10:41.000000000","tz":0},"subject":"setup.cfg: Remove unused \u0027build_sphinx\u0027 configuration","message":"setup.cfg: Remove unused \u0027build_sphinx\u0027 configuration\n\nWe call \u0027sphinx-build\u0027 directly now. No need to keep this legacy\nconfiguration around.\n\nChange-Id: I337f9fefe7f0364bce3a0192dfc9c1199076fa02\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/abba182e12889f162e2ad992aa66bd5dfb4a7fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/abba182e12889f162e2ad992aa66bd5dfb4a7fed"}]},"branch":"refs/heads/master"},"c4ee198eb7f95f7192d9d5dcb157289848bc2506":{"kind":"NO_CHANGE","_number":2,"created":"2020-02-27 15:05:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/704535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/35/704535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/704535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/704535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/35/704535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/35/704535/2"}}},"commit":{"parents":[{"commit":"e3e20d435e0f8cf887da1eb0497f66b5052274af","subject":"setup.cfg: Configure \u0027python-requires\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/e3e20d435e0f8cf887da1eb0497f66b5052274af"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-28 11:10:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-27 15:04:37.000000000","tz":0},"subject":"setup.cfg: Remove unused \u0027build_sphinx\u0027 configuration","message":"setup.cfg: Remove unused \u0027build_sphinx\u0027 configuration\n\nWe call \u0027sphinx-build\u0027 directly now. No need to keep this legacy\nconfiguration around.\n\nChange-Id: I337f9fefe7f0364bce3a0192dfc9c1199076fa02\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/c4ee198eb7f95f7192d9d5dcb157289848bc2506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/c4ee198eb7f95f7192d9d5dcb157289848bc2506"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
