)]}'
{"id":"openstack%2Fos-traits~710274","triplet_id":"openstack%2Fos-traits~master~Iac33e324d07d67368a5b544030cd5414d0586955","project":"openstack/os-traits","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Iac33e324d07d67368a5b544030cd5414d0586955","subject":"Remove use of six","status":"MERGED","created":"2020-02-27 14:25:50.000000000","updated":"2020-02-27 17:48:58.000000000","submitted":"2020-02-27 17:45:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710274-1582825513167-28c1a561","meta_rev_id":"59e11602c741357c79e5d10dab7706325ecdafa2","_number":710274,"virtual_id_number":710274,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 17:45:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2020-02-27 15:05:53.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2020-02-27 16:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2020-02-27 16:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 14:39:49.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-02-27 16:31:11.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-02-27 17:45:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf75b552117e157158dfc97403cb6b70810e36c9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-27 14:25:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70bb1aaae44ac79f6e1dbcbb5f5fe4b77ef9761d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-27 14:39:49.000000000","message":"Patch Set 1: Code-Review+2\n\neasy win","accounts_in_message":[],"_revision_number":1},{"id":"d4f631427cd35fceeac6b9c2d87df2e02a4d862d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:58:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be37628bde644226b10e6c9e1747f307 : SUCCESS in 9m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0258b80cfeab4243bdd04076c156c0d3 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a23ea2f6b94ee18e7aa6260e394312 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e1a0c9f72a041938c96512664b31960 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49772c29ede94ee1b416172feffdea25 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8c28d373de74a11a2f3c212b148fe32 : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3e9fea385f5a66dc3edf16b30b29f6b3a949be9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-27 15:05:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"479397a3f44cd80eace762e0898e73da7ccf6dfd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-27 16:31:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5297d48f8aa6fe2af1515c4a01040b897811605","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:31:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10a9c0cf8837546e2f244c4a6504c0aa099159cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:32:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf378461d80f49d48c5f0b4ee4115e76 : SUCCESS in 8m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/36dface9188d47e4825e32909c72c03e : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e6a845c4664edc895e52a355ca10d5 : POST_FAILURE in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0fc02a421cf44be864b11941f866b6b : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bf62f47ea6d4919927322dd8f93f17c : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93eaabd32bd34ad3a5fbe0df093d37ac : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"afb792b4ac942d30513c0ca04ff15939d82aad70","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-27 16:34:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"64589fd20cffe5657f39983d2814cccd16a59de0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:00:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a4bd88bbf14c2cbe094507b163ac16 : SUCCESS in 9m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f39df75e4f74570a94634214c6cf01c : POST_FAILURE in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc37392c05414300874c70aac68f74ba : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ed1e90318774b8eb0d892cd8145a378 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b8c261162d2430ba610cde791d20e83 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"8e3db293c1171f28f13e3e4af06ba90478513586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:10:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9f00a3bd179429ea771d68df0a72cf9 : SUCCESS in 9m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/30d12b69a1334542a73aa7f07397a432 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe2b2decadd4a95a31d58d2aa352a76 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12170a46a2f0437bb4c352c7ccfd57a4 : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b41264dcb72e419ea53203776d20cc83 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/763e5042a5994c889c83cebf58df65c6 : SUCCESS in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1fae4f55327625202e70f4d9c8d9957297ae1b4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:10:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fdc6827d1b456cb735d405f244b1c9a78185264d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:45:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2d95c1175a47f4ad8cbf862b7103ee : SUCCESS in 6m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8bea76cdc82749db9ada1a56b300fefc : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29ea4a96305d4e12afd5377a253c7b42 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1193445168e4d839ae86a0156e82253 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7598d71b4ea34b63b2d89f3ac81198a8 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":2},{"id":"33d527d936b08030c64c3171eaf246ffb294640d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:45:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"93fce09d11c384b43154b4da19bf72b5d35ecede","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:48:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a2900050e94f228b6f054035bd23b0 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"865cf20732383b494ebf410ec6b4b3577bb3dd20","revisions":{"5cfefcb0dc33918b917501cfbb8e38dbef41b979":{"kind":"REWORK","_number":1,"created":"2020-02-27 14:25:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/710274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/74/710274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/74/710274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/74/710274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/74/710274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/74/710274/1"}}},"commit":{"parents":[{"commit":"5b747f311e8deddd25f8b0cf428c6656a5acbf6d","subject":"setup.cfg: Configure \u0027python-requires\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/5b747f311e8deddd25f8b0cf428c6656a5acbf6d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-27 14:14:11.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-27 14:15:12.000000000","tz":0},"subject":"Remove use of six","message":"Remove use of six\n\nOnly one user here which is no longer needed in a Python 3 only world.\n\nChange-Id: Iac33e324d07d67368a5b544030cd5414d0586955\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/5cfefcb0dc33918b917501cfbb8e38dbef41b979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/5cfefcb0dc33918b917501cfbb8e38dbef41b979"}]},"branch":"refs/heads/master"},"865cf20732383b494ebf410ec6b4b3577bb3dd20":{"kind":"NO_CHANGE","_number":2,"created":"2020-02-27 15:05:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/710274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/74/710274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/74/710274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/74/710274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/74/710274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/74/710274/2"}}},"commit":{"parents":[{"commit":"c4ee198eb7f95f7192d9d5dcb157289848bc2506","subject":"setup.cfg: Remove unused \u0027build_sphinx\u0027 configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/c4ee198eb7f95f7192d9d5dcb157289848bc2506"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-27 14:14:11.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-27 15:04:41.000000000","tz":0},"subject":"Remove use of six","message":"Remove use of six\n\nOnly one user here which is no longer needed in a Python 3 only world.\n\nChange-Id: Iac33e324d07d67368a5b544030cd5414d0586955\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/865cf20732383b494ebf410ec6b4b3577bb3dd20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/865cf20732383b494ebf410ec6b4b3577bb3dd20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
