)]}'
{"id":"openstack%2Fos-traits~770530","triplet_id":"openstack%2Fos-traits~master~I0d467fe96f31a8d02c97b0555ae469faa879e67f","project":"openstack/os-traits","branch":"master","topic":"vhost-vdpa","hashtags":[],"change_id":"I0d467fe96f31a8d02c97b0555ae469faa879e67f","subject":"add vdpa trait","status":"ABANDONED","created":"2021-01-13 04:57:38.000000000","updated":"2021-02-11 17:38:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"52487b47603d65eb322a040f7fd05d65fbc756a8","_number":770530,"virtual_id_number":770530,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-13 05:21:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2021-02-11 17:08:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 05:08:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 17:08:08.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"940d2392b0dba2f552d18da81dce5cc65c313151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-13 04:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdc6d0ea41d88ac5e8faf14662a260ed7cc12c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 05:08:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383339d024a644f89f234486ecda02e1 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fff26b47bfd489391393d63a77f7dc7 : FAILURE in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb30c349643402abf6671fc3ac5035c : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8469caeeaed84995a65d1b13c20dd06c : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/498395638aae4fdb9e84ba88a7b70699 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c2f8b33bc93ef69c69e764f14109084bdd0a375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-01-13 05:12:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbd8b846e382b07eb957021aad4065c4bd13f4f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 05:21:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d1dbf6f2006493199727d5825af9100 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e983241a032e47b69818e61b49dd6541 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aca63c81a2a64ec9bc6df48926a57e46 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83aab7cba7b5419e80088a3f8f2dc2e6 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c758a538277467889f9754df6274409 : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c9a216967d1a76fd93f54b91e2cc3abb256b278","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-02-11 17:08:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"52487b47603d65eb322a040f7fd05d65fbc756a8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-02-11 17:38:09.000000000","message":"Abandoned\n\nthanks for the review gibi but we removed this right at the end of the spec reveiw https://specs.openstack.org/openstack/nova-specs/specs/wallaby/approved/libvirt-vdpa-support.html#alternatives\n\nill repoen this if we need it in the future but the plan is just to use the pci passthough filter","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ec5dc5b910b78bf2bf5eeb8de5e237a87d6c4fb","revisions":{"6261348ffe7fca4f69f3d35e47236c9b636e5993":{"kind":"REWORK","_number":1,"created":"2021-01-13 04:57:38.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/30/770530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/30/770530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/30/770530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/30/770530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/30/770530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/30/770530/1"}}},"commit":{"parents":[{"commit":"1b14058122aa62439fb23de6efb10cce4a3f0c84","subject":"Add contextual comment about image type traits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/1b14058122aa62439fb23de6efb10cce4a3f0c84"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-01-13 04:55:06.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-01-13 04:56:42.000000000","tz":0},"subject":"add vdpa trait","message":"add vdpa trait\n\nThis change adds a standard trait for vhost datapath\nacclearation HW_NIC_VDPA. This trait will be advertised\nby host that support neutron port of vnic_type vdpa.\n\nPartial-Implements: blueprint libvirt-vdpa-support\nChange-Id: I0d467fe96f31a8d02c97b0555ae469faa879e67f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/6261348ffe7fca4f69f3d35e47236c9b636e5993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/6261348ffe7fca4f69f3d35e47236c9b636e5993"}]},"branch":"refs/heads/master"},"4ec5dc5b910b78bf2bf5eeb8de5e237a87d6c4fb":{"kind":"REWORK","_number":2,"created":"2021-01-13 05:12:05.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/30/770530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/30/770530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/30/770530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/30/770530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/30/770530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/30/770530/2"}}},"commit":{"parents":[{"commit":"1b14058122aa62439fb23de6efb10cce4a3f0c84","subject":"Add contextual comment about image type traits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/1b14058122aa62439fb23de6efb10cce4a3f0c84"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-01-13 04:55:06.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2021-01-13 05:11:48.000000000","tz":0},"subject":"add vdpa trait","message":"add vdpa trait\n\nThis change adds a standard trait for vhost datapath\nacclearation HW_NIC_VDPA. This trait will be advertised\nby host that support neutron port of vnic_type vdpa.\n\nPartial-Implements: blueprint libvirt-vdpa-support\nChange-Id: I0d467fe96f31a8d02c97b0555ae469faa879e67f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/4ec5dc5b910b78bf2bf5eeb8de5e237a87d6c4fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/4ec5dc5b910b78bf2bf5eeb8de5e237a87d6c4fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
