)]}'
{"id":"openstack%2Fos-traits~770570","triplet_id":"openstack%2Fos-traits~master~I333147dcd47c6d0b926338a5a0c545f5adc63961","project":"openstack/os-traits","branch":"master","topic":"bp/allow-secure-boot-for-qemu-kvm-guests","hashtags":[],"change_id":"I333147dcd47c6d0b926338a5a0c545f5adc63961","subject":"Add a trait for UEFI Secure Boot support","status":"MERGED","created":"2021-01-13 10:34:44.000000000","updated":"2021-02-11 17:26:23.000000000","submitted":"2021-02-11 17:24:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770570-bp/allow-secure-boot-for-qemu-kvm-guests","meta_rev_id":"4cee4c9d56ae2525758452595163c2a03d2f78df","_number":770570,"virtual_id_number":770570,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-11 17:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-02-10 16:23:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-02-11 17:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-02-11 17:06:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 11:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 16:23:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-02-11 17:06:52.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"a10d550cb3604fd82599eb92eb2899c4906d6459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2021-01-13 10:34:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a1b52fb7fbfbe02c15d8e98db5d78b94f68f075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:08:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b12466f9f44ba1b22d70d5deb7e7e3 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f869086f2c46e2979f19e48c2b81f4 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a15493f89d2e4621beb26de64dbe4e2c : RETRY_LIMIT in 2m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3047f29111664b0387366302a5cda366 : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3934ab405fe7419bb5512fdd194f8bff : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d14dde1620ad58ae65ddd7fb56b6f716e834e950","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-10 16:23:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6e5be0e4c023d03d22e972f36983ff5a6f14877","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-02-11 17:06:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ncomment nit can be fixed separately.","accounts_in_message":[],"_revision_number":1},{"id":"59e283b584cad644cc2b6ec8b11ceef309ccc62a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:15:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56abe814a7af4d28bd053e3d323c5348 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2be11b1392840c68d342a89c13d4de1 : SUCCESS in 2m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05216107e0cc4f4f927ab4d958e2c571 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9487b8fcaf3247e5a06713bf1cd25867 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8f5f481d1d64dc49a1e1293baac7fce : SUCCESS in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2bcb22880462de76b7264c20f25f03bfa70a1ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:15:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"666cf8282cb5ede3c7ece04e1b02a5cce1eb5bde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:24:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d73036c2044f44b61c43ad2b8aa627 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed03978196141c7a0246b9ac4432924 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a65844a92de149d6a151cb06dadf3693 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/145e72737e6645f2bbe5ce08f06ef606 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3a93a0d38a7a5366943dfa0a8070fd49113b9ceb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:24:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4cee4c9d56ae2525758452595163c2a03d2f78df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:26:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a61dbeca94b54ec2a8aa8776ba21e472 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89cec2696b95f924e277dc1eda091a545e359fa6","revisions":{"89cec2696b95f924e277dc1eda091a545e359fa6":{"kind":"REWORK","_number":1,"created":"2021-01-13 10:34:44.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/70/770570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/70/770570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/70/770570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/70/770570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/70/770570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/70/770570/1"}}},"commit":{"parents":[{"commit":"454292dbd22edd2e9565bfe790b74b91c068f2e6","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/454292dbd22edd2e9565bfe790b74b91c068f2e6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2021-01-12 15:17:14.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2021-01-13 10:31:56.000000000","tz":60},"subject":"Add a trait for UEFI Secure Boot support","message":"Add a trait for UEFI Secure Boot support\n\nIntroduce a trait, COMPUTE_SECURITY_UEFI_SECURE_BOOT, as discussed in\nthe Secure Boot spec[1].  (Yeah, it reads a little awkward, not sure if\nwe can avoid it, as I\u0027ve included it as part of compute/security.py\nfile.)\n\nThis lets an admin set the trait either via the flavor extra-specs or\nimage metadata properties:\n\n    trait:COMPUTE_SECURITY_UEFI_SECURE_BOOT\u003drequired\n\nto request to boot an instance in Secure Boot mode.  Nova\u0027s libvirt\ndriver is responsible make sure the compute node has the necessary\ncapabilities (e.g. relevant libvirt, QEMU, OVMF, EDK2 et al versions).\nIf the host hypervisor is incapable, the instance creation will fail.\n\n[1] https://opendev.org/openstack/nova-specs/src/branch/master/specs/wallaby/approved/allow-secure-boot-for-qemu-kvm-guests.rst#fnref8\n\nblueprint: allow-secure-boot-for-qemu-kvm-guests\n\nChange-Id: I333147dcd47c6d0b926338a5a0c545f5adc63961\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/89cec2696b95f924e277dc1eda091a545e359fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/89cec2696b95f924e277dc1eda091a545e359fa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
