)]}'
{"id":"openstack%2Fos-traits~779495","triplet_id":"openstack%2Fos-traits~master~Ifc06d6f9b1d3217d09fa20aafc2988acb11c80f2","project":"openstack/os-traits","branch":"master","hashtags":[],"change_id":"Ifc06d6f9b1d3217d09fa20aafc2988acb11c80f2","subject":"Add \u0027COMPUTE_FIRMWARE_UEFI\u0027, \u0027COMPUTE_FIRMWARE_BIOS\u0027","status":"MERGED","created":"2021-03-09 12:47:22.000000000","updated":"2021-08-19 14:33:03.000000000","submitted":"2021-08-19 14:31:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779495","meta_rev_id":"86f9ffc2c4d6f83ba872e0fee6ce1c76107af6bc","_number":779495,"virtual_id_number":779495,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 14:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-08-19 11:45:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-08-19 14:20:02.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-19 10:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2021-08-19 11:50:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2021-08-19 14:20:02.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 13:02:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-17 14:05:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-08-17 16:38:41.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-08-19 11:50:47.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-08-19 14:20:02.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"}],"messages":[{"id":"6685097718d2b7139161678b6bbe6ef43adfb926","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-09 12:47:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58e074aedc3e6948a8d1359943d0ce4963d54da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 13:02:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ccd38e502e4cc98499f8089961f503 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db252214ce814b0d8ccd1460b398ecb8 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d3a96459fa843878c73ccf18fce5212 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c4b71559fd444eeb1ba4a850771504d : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6716642495a94e7f96bb6f1e98977d1f : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"748906f146174c509c0a77261b66ba27598d2579","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-17 14:05:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bacdf7098ab25bfd3e76094be85495231946171f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-17 16:31:43.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"5426c9f3f9f4c86ad02cb440460bfe5a360a6fe2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-17 16:38:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1145155f6b30342541953884da0436d10a77c3e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-19 09:54:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d736e89b0b667bee6a772e0fc8a66412b0573b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-19 09:56:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e08ada0b3fd580add6b514533228890e8ec8403e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:07:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b198455e21414f95c4d626ff94669b : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23b9c86b396f40b188ed9fb7ab05ef29 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0833ed65a7e347d4862d16acdb375d73 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c7049eeaf26426d8271ea1023a7a76f : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a23f02fd8ac4ed7a3249ee9ec96dec4 : SUCCESS in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f132f6f569df9af6c79baef27ede4860e5fb0fe","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-19 10:35:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e05a685657bfb3a4ca8e416d835039e370d384f5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-19 11:45:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7b455cd7359ecd423b47913efe83ef93c9d156d3","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-19 11:50:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0620c5923cf2cd9f5bb01ebd727b36c8799b41d4","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-08-19 14:20:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e343fb513f392bf977e22509ed060c2d1e747567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:20:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83ef36fad3af3f498ef63b1a59be90e4499a7d06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:31:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/384bb67b6cca48ac8818fe24984ac0ed : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769e2633ca6249c384ace39b6f269e17 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512091f0fef2419babf5d2f448a2404e : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ee693fd77fa44349871fcf3dd790f03 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"397147f024210db9896516f7b65330f15a7faa6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:31:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"86f9ffc2c4d6f83ba872e0fee6ce1c76107af6bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:33:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b7a7213ec6413cb8b5fd47a7ad612f : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32112e9091caf6e87552c933c5607b6548d5e294","revisions":{"ea260110175728565afd064c7ba1a6d550b9fc48":{"kind":"REWORK","_number":1,"created":"2021-03-09 12:47:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/779495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/95/779495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/95/779495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/95/779495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/95/779495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/95/779495/1"}}},"commit":{"parents":[{"commit":"ac1b39eff3e61b0e95cffad4510f71b6b2745c4b","subject":"Merge \"Add COMPUTE_EPHEMERAL_ENCRYPTION tratis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/ac1b39eff3e61b0e95cffad4510f71b6b2745c4b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-03-09 12:42:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-03-09 12:45:40.000000000","tz":0},"subject":"Add \u0027HW_FIRMWARE_UEFI\u0027, \u0027HW_FIRMWARE_BIOS\u0027","message":"Add \u0027HW_FIRMWARE_UEFI\u0027, \u0027HW_FIRMWARE_BIOS\u0027\n\nThis allows us to report this trait and use it to filter out hosts that\nsupport our requested \u0027hw_firmware_type\u0027.\n\nChange-Id: Ifc06d6f9b1d3217d09fa20aafc2988acb11c80f2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/ea260110175728565afd064c7ba1a6d550b9fc48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/ea260110175728565afd064c7ba1a6d550b9fc48"}]},"branch":"refs/heads/master"},"32112e9091caf6e87552c933c5607b6548d5e294":{"kind":"REWORK","_number":2,"created":"2021-08-19 09:54:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/779495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/95/779495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/95/779495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/95/779495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/95/779495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/95/779495/2"}}},"commit":{"parents":[{"commit":"ac1b39eff3e61b0e95cffad4510f71b6b2745c4b","subject":"Merge \"Add COMPUTE_EPHEMERAL_ENCRYPTION tratis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/ac1b39eff3e61b0e95cffad4510f71b6b2745c4b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-03-09 12:42:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-08-19 09:54:12.000000000","tz":60},"subject":"Add \u0027COMPUTE_FIRMWARE_UEFI\u0027, \u0027COMPUTE_FIRMWARE_BIOS\u0027","message":"Add \u0027COMPUTE_FIRMWARE_UEFI\u0027, \u0027COMPUTE_FIRMWARE_BIOS\u0027\n\nThis allows us to report this trait and use it to filter out hosts that\nsupport our requested \u0027hw_firmware_type\u0027.\n\nChange-Id: Ifc06d6f9b1d3217d09fa20aafc2988acb11c80f2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/32112e9091caf6e87552c933c5607b6548d5e294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/32112e9091caf6e87552c933c5607b6548d5e294"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
