)]}'
{"id":"openstack%2Fos-traits~791973","triplet_id":"openstack%2Fos-traits~master~Id377e233879b87e26ba92ff667b98cc0c1210db2","project":"openstack/os-traits","branch":"master","topic":"update-min-tox","hashtags":[],"change_id":"Id377e233879b87e26ba92ff667b98cc0c1210db2","subject":"Change minversion of tox to 3.18.0","status":"MERGED","created":"2021-05-18 13:49:00.000000000","updated":"2021-10-01 10:11:31.000000000","submitted":"2021-10-01 10:10:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791973-update-min-tox","meta_rev_id":"857ac4808d81c3554f3067afa81475d9ed2720dc","_number":791973,"virtual_id_number":791973,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-01 10:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-10-01 10:11:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-10-01 10:01:39.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-10-01 10:01:39.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 13:56:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 13:33:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-08-20 15:11:47.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"}],"messages":[{"id":"593bb3ab119d42efd9905cd690779694a0d9185c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-05-18 13:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2ef46bb8d5de3e762e71beb57d5a0739b873f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:56:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/377d538274684a558b7fdade1a41e792 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50abe9a61904442eb39403bee7a371b8 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ece2cf894dd5400bacb1663a0382fd4e : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df8d43e078a34d30bdacb843a5a76f5a : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7ffa77111c441e9ae2aefec0579b6ae : SUCCESS in 4m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f717ffba3494a628f49921f772cb44c : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2d665ac1e1758e5458e6bcc0616fc71b0f3d76b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-05-18 14:02:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"27796c3df4c79979be4e82402a4b130cbe3f2b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 14:14:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed643476fb3436e8f2f4803c7bf0f9d : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d5a241263cd45ca940e5b434fe2be61 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54d245e1525743d49bdf9bb3b13e670a : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7701e405fab4c3c864c26b7efff4ce1 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80f968f3b805479a81d321e16aa10b60 : SUCCESS in 3m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/009ab355f6bf4f97a0b1e00ba422be21 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"cf6713c83f25af7d1460572081430c02156bad06","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-01 13:33:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nTrivial, test-only change. Fast approving.","accounts_in_message":[],"_revision_number":2},{"id":"efc266fb16310e3c9960ac74853d45829082723b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 13:33:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d09377a6198722c251badc96561fb81a8f816c47","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-01 13:34:21.000000000","message":"Patch Set 2: -Workflow\n\nAh, I see Sean has concerns on the os-vif change. Perhaps this is more controversial than I thought","accounts_in_message":[],"_revision_number":2},{"id":"aa374b9794085996edb90e7ac075bb16232129ce","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-01 13:38:06.000000000","message":"Patch Set 2:\n\nfwiw, Sean\u0027s concerns are that this will force users on CentOS 8 and Ubuntu 20.04 to get tox from PyPI as opposed to the distro\u0027s repos. I think that\u0027s an acceptable ask but I\u0027m also biased as I use Fedora. Will let someone else that uses these distros decide if they want to block this","accounts_in_message":[],"_revision_number":2},{"id":"fd58ac2e9957916c68d09bf300277e8828d2b4fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 13:43:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0324ceb500e541329699e4ecee948265 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a609e7924ea74f1a845eae89659abd00 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/417f6ed5cb194648817fb2de32a226c5 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34ea1f6d8e554a41b58fe0d1526301d9 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2de6a2aa24f94b628bb240fb06adb79a : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"89922364aad51672876205c74927b276df9d0d04","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-08-20 15:11:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0bad5c75245fdd4cb9e5e6e5c314b73dac5e5659","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-08-20 15:12:09.000000000","message":"Removed Workflow+1 by Lee Yarwood \u003clyarwood@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"23e70151bdd5a083442bd6b687bbe6f8bc69db16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 15:12:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df4678c4fc27f00fade2ab22597f008e5b6b56a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 15:23:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c0214bc6cc4b61bd5362bf13044278 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12b83c44e35444f9e53992248382b2b : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/088cb91a241046d2ab835b2ba14eaf97 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc8fd7ae0406458d8b3e90be4509ab08 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d482c28186a49699d22f4bb2808b7e9 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d8f9c186a6da02eb6595257c103e9a92ee7b7bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-09-04 02:17:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b87d184a314cd8f68cc1181090400d44a6499e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 02:27:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cfcfc6e98424c36add33573694f954c : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e18b91424ae4371822018d68378f275 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/960b67de441e48da9ee40c67cbd27b3c : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b0828a0467047988ef9ad3fdbb8f68d : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99372489c23e4ddb890e3c622beaefff : SUCCESS in 3m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1566f71a155442fa8bdddec6569e4450 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b574e4b68cd7496118d018019cc14a1d2581e7ec","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-10-01 10:01:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"37a18569e3ae90838493cb25c16b7be3ef55f012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 10:02:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9645988025802da99d83e1c38f7a51eb943c10c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 10:10:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a82c47f25744659f86e0361559f11e : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea1379506a745d4ba182f51fe385621 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bb70d6290f14e5cb818ee3f627719a8 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0836eb1aefd4fb0a58c98c42cbd50e5 : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b5994a300b94dcc83827383112a1b38 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"92f877b5d4ae565e2616cd485ead6bfb4d728eb9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 10:10:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0e96c35e739b196b67deb4416217b65792fdb0d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 10:11:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c734ef3307084139a907fae9d82a0634 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e91acd7e12234b0e9120fa129ea9ff7e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ff96dcc64767cc47e2a736d925c3fd31c9929262","revisions":{"fda26658fc033c7891c74310b4ef5d733f3975e9":{"kind":"REWORK","_number":1,"created":"2021-05-18 13:49:00.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/73/791973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/73/791973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/73/791973/1"}}},"commit":{"parents":[{"commit":"ac1b39eff3e61b0e95cffad4510f71b6b2745c4b","subject":"Merge \"Add COMPUTE_EPHEMERAL_ENCRYPTION tratis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/ac1b39eff3e61b0e95cffad4510f71b6b2745c4b"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:38:36.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:38:36.000000000","tz":540},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 python in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: Id377e233879b87e26ba92ff667b98cc0c1210db2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/fda26658fc033c7891c74310b4ef5d733f3975e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/fda26658fc033c7891c74310b4ef5d733f3975e9"}]},"branch":"refs/heads/master"},"b4de4e7c124c677edaff1e8a942c2da2c80d1e64":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-18 14:02:18.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/73/791973/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/73/791973/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/73/791973/2"}}},"commit":{"parents":[{"commit":"ac1b39eff3e61b0e95cffad4510f71b6b2745c4b","subject":"Merge \"Add COMPUTE_EPHEMERAL_ENCRYPTION tratis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/ac1b39eff3e61b0e95cffad4510f71b6b2745c4b"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:38:36.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 14:01:47.000000000","tz":540},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: Id377e233879b87e26ba92ff667b98cc0c1210db2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/b4de4e7c124c677edaff1e8a942c2da2c80d1e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/b4de4e7c124c677edaff1e8a942c2da2c80d1e64"}]},"branch":"refs/heads/master"},"ff96dcc64767cc47e2a736d925c3fd31c9929262":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-04 02:17:10.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/73/791973/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/73/791973/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/73/791973/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/73/791973/3"}}},"commit":{"parents":[{"commit":"c323982d9ec0cb387bde6ea676a44687eb56b49f","subject":"Merge \"List traits in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/c323982d9ec0cb387bde6ea676a44687eb56b49f"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:38:36.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-09-04 02:17:10.000000000","tz":0},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: Id377e233879b87e26ba92ff667b98cc0c1210db2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/ff96dcc64767cc47e2a736d925c3fd31c9929262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/ff96dcc64767cc47e2a736d925c3fd31c9929262"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
