)]}'
{"id":"openstack%2Fos-traits~844336","triplet_id":"openstack%2Fos-traits~master~Ic44429ad360475143b20071f85d534a958d110e0","project":"openstack/os-traits","branch":"master","topic":"bp/virtual-iommu-support","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-07-28 14:26:56.000000000","reason":"Change was submitted"},"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2022-07-28 14:26:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic44429ad360475143b20071f85d534a958d110e0","subject":"Add traits for vIOMMU","status":"MERGED","created":"2022-06-01 20:34:25.000000000","updated":"2022-07-28 14:29:02.000000000","submitted":"2022-07-28 14:26:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"844336-bp/virtual-iommu-support","meta_rev_id":"2412a84900ea61339f5104652952768fa53b2d46","_number":844336,"virtual_id_number":844336,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 14:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-07-28 11:27:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-28 14:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-28 14:15:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 20:48:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 11:31:16.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-07-28 11:27:31.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-07-28 14:15:52.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"35515160ff85d6c8ec8634a8aa1cb3e144d3f570","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2022-06-01 20:34:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82566f0441f7719e0fdea49c6cb9552c17bc446e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 20:48:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ebcd8b4bc34674a151a8a3f5e30c87 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22f045bcf7a846f2bcefbc719b2e774f : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65e15be1590141178ff900370db66849 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6428d024dad94aff9ed947a880c638d7 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c96644e9ec5d45f29bc5a782eedf72c5 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7baeb1f96ed6a627019a14bb91fd8db77bfb07b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-18 11:31:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c6936ae82eb5362dcce0ab2625667a40ed76dd99","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-18 11:31:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21abcf215b4f1e937825d0c1a77d5309a63b035b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-18 11:38:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"576234091cb9f81492da35e3e6d63b813b0df9fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-28 11:21:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"71ad983f227a69396c07c815e904db8d36a8532d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 11:21:50.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"54b7c8a1f9005a31ef438164fdd8046ea3cb643b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-28 11:22:16.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dd3910f3a5d8d8134264753420168777414984c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-28 11:22:48.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7685cdced561495e1633ce46abe626fb55b19eac","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-07-28 11:27:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fde8aa834eba2ca26b15dd9879d8250317d127e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 11:30:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5900da1d574df2bc4bcb0058158084\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16ec391a50e647738eb84d19dc9c8f81 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08bd52d62cf1464cba9cfb0ce2a38021 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d82c0030b8cb4af2be6cc4ce7504fe65 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dde11a411e041c8b9c282f9c275ff0c : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a274f19880e44cd935f5b79bc4e52dd : SUCCESS in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e51c1002828e6226d5637f4a7be947eba0f8ed2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-28 14:15:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d12f600e0ebc0f636a9665198d0718cf8926dbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 14:16:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"158f6b980c637b796aca34a7ba88d6c533a8dfdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 14:26:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be5a0204e593466abaaaebde85cab6a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c4ef288b8442f2b1e1572b32d606fb : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e15077581864d3e8b9c4b5a658ab9cc : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e1b3b1d5af54eb09580444b5eadf467 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb2c39f9a90d40d392623e1408b37105 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":3},{"id":"ac31fcb8922d22042c1820e0dd6d4177edea710e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 14:26:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2412a84900ea61339f5104652952768fa53b2d46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 14:29:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65f12c96e7ba43349c011ba8638d01b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0154013a036b41a497ee7659f974b2eb : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"74988cdba731d9b48b15b76372da8460304ea937","revisions":{"aba0bd8ef3acf9736e230c120da5c5cee85a23bd":{"kind":"REWORK","_number":1,"created":"2022-06-01 20:34:25.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/36/844336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/36/844336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/36/844336/1"}}},"commit":{"parents":[{"commit":"f64d50e4dd2f21558fb73dd4b59cd1d4b121b707","subject":"Add nova and cyborg owner traits for resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/f64d50e4dd2f21558fb73dd4b59cd1d4b121b707"}]}],"author":{"name":"Rico Lin","email":"rlin@vexxhost.com","date":"2022-06-01 20:28:52.000000000","tz":480},"committer":{"name":"Rico Lin","email":"rlin@vexxhost.com","date":"2022-06-01 20:34:21.000000000","tz":480},"subject":"Add traits for vIOMMU","message":"Add traits for vIOMMU\n\nAdd traits for vIOMMU traits corresponding to the allowed values of\n\"hw_iommu_model\" image metadata property.\n\nBlueprint: libvirt-viommu-device\nDepends-On: https://review.opendev.org/c/openstack/nova/+/830646\nChange-Id: Ic44429ad360475143b20071f85d534a958d110e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/aba0bd8ef3acf9736e230c120da5c5cee85a23bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/aba0bd8ef3acf9736e230c120da5c5cee85a23bd"}]},"branch":"refs/heads/master"},"abfa82dd65a0dc104c51660155d3a41539f6c118":{"kind":"REWORK","_number":2,"created":"2022-07-28 11:21:25.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/36/844336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/36/844336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/36/844336/2"}}},"commit":{"parents":[{"commit":"f64d50e4dd2f21558fb73dd4b59cd1d4b121b707","subject":"Add nova and cyborg owner traits for resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/f64d50e4dd2f21558fb73dd4b59cd1d4b121b707"}]}],"author":{"name":"Rico Lin","email":"rlin@vexxhost.com","date":"2022-06-01 20:28:52.000000000","tz":480},"committer":{"name":"sean mooney","email":"smooney@redhat.com","date":"2022-07-28 11:21:25.000000000","tz":0},"subject":"Add traits for vIOMMU","message":"Add traits for vIOMMU\n\nAdd traits for vIOMMU traits corresponding to the allowed values of\n\"hw_iommu_model\" image metadata property.\n\nBlueprint: libvirt-viommu-device\nDepends-On: https://review.opendev.org/c/openstack/nova/+/830646\nChange-Id: Ic44429ad360475143b20071f85d534a958d110e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/abfa82dd65a0dc104c51660155d3a41539f6c118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/abfa82dd65a0dc104c51660155d3a41539f6c118"}]},"branch":"refs/heads/master"},"74988cdba731d9b48b15b76372da8460304ea937":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-28 11:22:48.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/36/844336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/36/844336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/36/844336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/36/844336/3"}}},"commit":{"parents":[{"commit":"f64d50e4dd2f21558fb73dd4b59cd1d4b121b707","subject":"Add nova and cyborg owner traits for resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/f64d50e4dd2f21558fb73dd4b59cd1d4b121b707"}]}],"author":{"name":"Rico Lin","email":"rlin@vexxhost.com","date":"2022-06-01 20:28:52.000000000","tz":480},"committer":{"name":"sean mooney","email":"smooney@redhat.com","date":"2022-07-28 11:22:48.000000000","tz":0},"subject":"Add traits for vIOMMU","message":"Add traits for vIOMMU\n\nAdd traits for vIOMMU traits corresponding to the allowed values of\n\"hw_iommu_model\" image metadata property.\n\nBlueprint: libvirt-viommu-device\nChange-Id: Ic44429ad360475143b20071f85d534a958d110e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/74988cdba731d9b48b15b76372da8460304ea937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/74988cdba731d9b48b15b76372da8460304ea937"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
