)]}'
{"id":"openstack%2Fos-traits~868149","triplet_id":"openstack%2Fos-traits~master~I6605af9474a588ad283e5b1f571839c239e3c746","project":"openstack/os-traits","branch":"master","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-02-09 17:23:44.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"12597":{"account":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"last_update":"2026-02-09 17:34:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6605af9474a588ad283e5b1f571839c239e3c746","subject":"CPU: add traits for new X86 feature \"AMX\"","status":"MERGED","created":"2022-12-20 01:26:18.000000000","updated":"2026-02-09 17:35:22.000000000","submitted":"2026-02-09 17:34:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"868149","meta_rev_id":"0d1b496c711557fcd963cdf1e9111752b67a8fbb","_number":868149,"virtual_id_number":868149,"owner":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 17:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-02-09 17:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-09 16:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-02-09 17:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 01:31:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-24 22:25:03.000000000","updated_by":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-02-09 16:52:48.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"e6ee7f0b21b6f7fc60e57e1be7b3631daf61d2c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"date":"2022-12-20 01:26:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cb8c8d8169d2785f2b6f5b699b7c9490ccc59ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 01:31:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10c833497b7f44a897c89ba80127027a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4b18370a69419aa7baefa9eafb4a55 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3deceb3dac84d3b93c7bd431a3df8aa : SUCCESS in 2m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/299193bbd0c849359f7d08465c8aa4f7 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d5e3f6a44fa455ca3fceb3c84a80681 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"84887916abae4fbb92e2150241d3137d36872714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"date":"2023-04-24 21:59:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ceb117711c000bc6834fc83af6df37b23b22062b","author":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"date":"2023-04-24 22:03:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7b02e080ec28d71fc056c8c408203eb91c5a16e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 22:05:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d4e29b3e5ff44988f50334eb30f8719\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7352f7ca3b61440ca5fd7e9acf3c95af : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5db6c52011d4cba89749602489abad3 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c507d3c906b84b76bcd5fec2e1fbd890 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e6f33962cdc45ae9209a3e2beabc1cb : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"528e7b5d3b89dff23ffacc2809b2a69fdefe7b24","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-04-25 12:09:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fbf076fc6d8b3de470e4cd9f554cc9d5e43a15d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-04-25 12:14:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c71564ed6a73ef597f89e186eda3fe7457337ce1","author":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"date":"2023-05-18 03:56:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d013550f2151b6753a9248a844012175de6624b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"real_author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-09 16:51:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_12597\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"}],"_revision_number":3},{"id":"95229638aefd0666c9d5545243cd340e52e50734","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-02-09 16:52:48.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81179ea92eba0ae889975dfd76f975a3fc43ea2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:05:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f29a9a0805fc473895b9cc82142c041f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/725d722b93ce449e821d4e24432a34d9 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2b84024b4c41e682fc563225c78621 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daaba8c618ae4b3e9487181058bb29dc : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8fff32488f34674bb79574eec2e610d : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"59d1536d554c3eaf76d7f04389d1a840d47e04fe","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-09 17:23:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d8f32531e3ce08c02445bd57e8e96e344be00a8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:23:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3eee56722b4a849af046553cfed683a38b38320c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:34:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d756fb950912410b8ef6e58a7f856e70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e612922b40cd40c08d547fe89bb6e454 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b7349eb6d54db7b8c2d910c7810f4b : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d63e2cfcfe9d48f9a021cf30b20b9dd3 : SUCCESS in 1m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f8a1eb1b5474527be04692fd79c80b1 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"502550b2661225308b66aef82e3fa4db6cc1a848","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:34:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0d1b496c711557fcd963cdf1e9111752b67a8fbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:35:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b72bbf01d6bc4d28ac004ed17c3d4c09\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a452a5f9c74414db0a9e788c0fc196d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"15d22a70665039c89b9899606da91d1337dd2e14","revisions":{"1ed7094e7dcc90b52418f2f622fa8fda3f3eeab3":{"kind":"REWORK","_number":1,"created":"2022-12-20 01:26:18.000000000","uploader":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"ref":"refs/changes/49/868149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/49/868149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/49/868149/1"}}},"commit":{"parents":[{"commit":"2dbc9721b0e3c24c41f6f2a4703bc29b6b778af8","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/2dbc9721b0e3c24c41f6f2a4703bc29b6b778af8"}]}],"author":{"name":"Lin Yang","email":"lin.a.yang@intel.com","date":"2022-12-20 01:17:06.000000000","tz":-480},"committer":{"name":"Lin Yang","email":"lin.a.yang@intel.com","date":"2022-12-20 01:19:55.000000000","tz":-480},"subject":"CPU: add traits for new X86 feature \"AMX\"","message":"CPU: add traits for new X86 feature \"AMX\"\n\nAdd three standard traits for the support of new CPU AMX [1]\ninstructions, which includes \u0027AMXBF16\u0027, \u0027AMXINT8\u0027 and\n\u0027AMXTILE\u0027.\n\n[1] https://en.wikipedia.org/wiki/Advanced_Matrix_Extensions\n\nSigned-off-by: Lin Yang \u003clin.a.yang@intel.com\u003e\nChange-Id: I6605af9474a588ad283e5b1f571839c239e3c746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/1ed7094e7dcc90b52418f2f622fa8fda3f3eeab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/1ed7094e7dcc90b52418f2f622fa8fda3f3eeab3"}]},"branch":"refs/heads/master"},"57e4943e93a795eff2e62ea0eea0dff706be1349":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-24 21:59:11.000000000","uploader":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"ref":"refs/changes/49/868149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/49/868149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/49/868149/2"}}},"commit":{"parents":[{"commit":"6c9917a7276c27197592d2583cd0719e7ea74f9e","subject":"Merge \"Update python testing as per zed cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/6c9917a7276c27197592d2583cd0719e7ea74f9e"}]}],"author":{"name":"Lin Yang","email":"lin.a.yang@intel.com","date":"2022-12-20 01:17:06.000000000","tz":-480},"committer":{"name":"Lin Yang","email":"lin.a.yang@intel.com","date":"2023-04-24 21:59:11.000000000","tz":0},"subject":"CPU: add traits for new X86 feature \"AMX\"","message":"CPU: add traits for new X86 feature \"AMX\"\n\nAdd three standard traits for the support of new CPU AMX [1]\ninstructions, which includes \u0027AMXBF16\u0027, \u0027AMXINT8\u0027 and\n\u0027AMXTILE\u0027.\n\n[1] https://en.wikipedia.org/wiki/Advanced_Matrix_Extensions\n\nSigned-off-by: Lin Yang \u003clin.a.yang@intel.com\u003e\nChange-Id: I6605af9474a588ad283e5b1f571839c239e3c746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/57e4943e93a795eff2e62ea0eea0dff706be1349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/57e4943e93a795eff2e62ea0eea0dff706be1349"}]},"branch":"refs/heads/master","description":"Rebase"},"15d22a70665039c89b9899606da91d1337dd2e14":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-02-09 16:51:19.000000000","uploader":{"_account_id":12597,"name":"Lin Yang","email":"lin.a.yang@intel.com","username":"lin.yang"},"real_uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/49/868149/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-traits","ref":"refs/changes/49/868149/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-traits refs/changes/49/868149/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-traits refs/changes/49/868149/3"}}},"commit":{"parents":[{"commit":"0715f71dfa43e12df8e6137d96736854ffba28e8","subject":"Drop basepython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/0715f71dfa43e12df8e6137d96736854ffba28e8"}]}],"author":{"name":"Lin Yang","email":"lin.a.yang@intel.com","date":"2022-12-20 01:17:06.000000000","tz":-480},"committer":{"name":"Lin Yang","email":"lin.a.yang@intel.com","date":"2026-02-09 16:51:19.000000000","tz":0},"subject":"CPU: add traits for new X86 feature \"AMX\"","message":"CPU: add traits for new X86 feature \"AMX\"\n\nAdd three standard traits for the support of new CPU AMX [1]\ninstructions, which includes \u0027AMXBF16\u0027, \u0027AMXINT8\u0027 and\n\u0027AMXTILE\u0027.\n\n[1] https://en.wikipedia.org/wiki/Advanced_Matrix_Extensions\n\nSigned-off-by: Lin Yang \u003clin.a.yang@intel.com\u003e\nChange-Id: I6605af9474a588ad283e5b1f571839c239e3c746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/15d22a70665039c89b9899606da91d1337dd2e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-traits/commit/15d22a70665039c89b9899606da91d1337dd2e14"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
