)]}'
{"id":"openstack%2Fos-vif~1007992","triplet_id":"openstack%2Fos-vif~master~Iea51c602d17726e394b07d338ef752153fbad8df","project":"openstack/os-vif","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-30 10:45:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea51c602d17726e394b07d338ef752153fbad8df","subject":"Fix mypy errors","status":"MERGED","created":"2026-09-30 01:03:38.000000000","updated":"2026-09-30 10:47:46.000000000","submitted":"2026-09-30 10:45:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007992","meta_rev_id":"475a9e8bfbf9aaca5edb0abd35e708f5af2be5a8","_number":1007992,"virtual_id_number":1007992,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-30 10:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-30 06:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-09-30 09:19:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-30 09:19:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-30 02:12:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-30 06:13:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-09-30 06:14:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-30 09:19:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b39c837ebdbfe16f39d4828869767d6a2f58ed12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-30 01:03:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81b5172b2adabf3d5f5bca5abbcc00b92e2bcc06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 02:12:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e06521cc7064befa4dac4b8782ec3df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbb429f399d4ecabbd7cadb6e35c893 : SUCCESS in 2m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51a97273843c481eb2d5f8e813b255c1 : SUCCESS in 2m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c792b6efaa64fb286ae84bd9b168eee : SUCCESS in 5m 05s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/fb262bd398d44a8cb06d612be78709a2 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c0824dd31240f5a128c5c260b7bbbe : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/606a68d225094cd896281756bde37305 : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/478b4852e426424bbd8772abb2a8abc2 : SUCCESS in 3m 53s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/3f62e10effc84425b524f1dee88320cc : POST_FAILURE in 2m 51s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/c56406e87d594ddfa5f04c5f9d197f63 : SUCCESS in 32m 45s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/7e1a0397f2f6442a8c41e7495a424df6 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"82be589a66400ca98aaeae05a03a9160480e16ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-30 06:13:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14ea13d57f44aec7d6421be8fefe426066d31f37","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-30 06:14:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c712113bfad160e204641cdbf58fb95f41e6bee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 07:34:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff143e17c394db791be7ba6d12b6450\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90695048efbd40b09f4a22aacdf7744d : SUCCESS in 2m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9776366c8d1e4adbb09fde4731addc83 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/648078cfbba54f4a9a6ffa8da7aa7cad : SUCCESS in 3m 22s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3dbb5ff0bd7645d2b535c4d374959e67 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ab6a456efe749adbde374d9db8cd475 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/879ea9c4123f4b63a8dd1bd4317b4bc3 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33af8087d05e4e459089c11ae239e331 : SUCCESS in 3m 30s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/5883d28229b44263992f8f0cf7172241 : SUCCESS in 3m 25s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/58cd5a3168004aa6b9f5b7861dc18653 : SUCCESS in 1h 13m 16s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/fd6dedb1988944fe8d4bf1e9d7f2e150 : SUCCESS in 1h 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e410e46f29dbd386232c445454cd22e667d05305","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-30 09:19:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"058462e6f36b153b3d0b67db0f2fa2e1d0b26ca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 09:20:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c6b10869d4e8a22fd9795ac637bf50eb5ed7401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 10:45:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f53f6ec78d54466b1373a8ec9957fbf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cce727b655e47459392baf84ac36904 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/595a4ec955ba4b0e94bd47f90f85f31c : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be4ff88d93b04096be3e2215be435427 : SUCCESS in 5m 18s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/85a13a623bef41a9b64fb359ab175b56 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56df22cddd5f4cb79c130cde50f20779 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c2818bb937f4c0f80da655a4c30f947 : SUCCESS in 3m 31s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/7a48606a1245407a8a0425e6d3075047 : SUCCESS in 4m 30s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/d518a23bc7bb4df08e2ddeefae5917ad : SUCCESS in 1h 20m 42s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/15812d06f684420da768a1754e1214ba : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1744c20895472f0d95f1d716fea61c79e4c551cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 10:45:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"475a9e8bfbf9aaca5edb0abd35e708f5af2be5a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 10:47:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4523b1e9723f4accb61030abd8f732f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770e5e91aad34324a938eeb0b4af59d8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fe3978d990aa7d16325b77f21ae2b22f2b6a567","revisions":{"4fe3978d990aa7d16325b77f21ae2b22f2b6a567":{"kind":"REWORK","_number":1,"created":"2026-09-30 01:03:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/1007992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/92/1007992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/92/1007992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/92/1007992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/92/1007992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/92/1007992/1"}}},"commit":{"parents":[{"commit":"70e9b383beb478620101d575344c6755302fe333","subject":"ovs: disable IPv6 on taps created by create_tap before bringing them up","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/70e9b383beb478620101d575344c6755302fe333"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-30 01:02:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-30 01:03:29.000000000","tz":540},"subject":"Fix mypy errors","message":"Fix mypy errors\n\nNote that the misc error is disabled because the failure is caused by\nthe typing definitions in testscenarios.\n\nerror: Definition of \"run\" in base class \"WithScenarios\" is\nincompatible with definition in base class \"TestCase\"  [misc]\n\nChange-Id: Iea51c602d17726e394b07d338ef752153fbad8df\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/4fe3978d990aa7d16325b77f21ae2b22f2b6a567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/4fe3978d990aa7d16325b77f21ae2b22f2b6a567"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
