)]}'
{"id":"openstack%2Fos-vif~636821","triplet_id":"openstack%2Fos-vif~master~I8308e8840e20b0a72d00880c1a7996b4c73f6a83","project":"openstack/os-vif","branch":"master","topic":"bug/1801919","hashtags":[],"change_id":"I8308e8840e20b0a72d00880c1a7996b4c73f6a83","subject":"remove brctl from vif_plug_ovs","status":"MERGED","created":"2019-02-14 04:07:43.000000000","updated":"2019-02-25 13:02:58.000000000","submitted":"2019-02-25 13:02:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636821-1551099778172-9837e355","meta_rev_id":"7011b5ae344b045a8f33b8ee50f818a4df066fab","_number":636821,"virtual_id_number":636821,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":2,"date":"2019-02-25 13:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-02-21 15:55:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":2,"date":"2019-02-25 09:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-02-25 12:50:42.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-02-25 09:06:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":1,"date":"2019-02-25 11:15:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-19 16:58:59.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-19 18:25:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-02-21 11:45:27.000000000","updated_by":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"reviewer":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"state":"REVIEWER"},{"updated":"2019-02-21 15:55:19.000000000","updated_by":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"reviewer":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"state":"REVIEWER"},{"updated":"2019-02-25 09:06:25.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2019-02-25 09:15:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-02-25 12:50:42.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2019-02-25 13:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c03a6924cc0e3c2bb9acaf508fc895464a8effb0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-14 04:07:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94f4d1cbb3cbaeadba03efa568b11934d0c271e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 06:14:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/636821/1/check/openstack-tox-lower-constraints/0f7469a/ : FAILURE in 6m 29s\n- openstack-tox-pep8 http://logs.openstack.org/21/636821/1/check/openstack-tox-pep8/feb2b3a/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/21/636821/1/check/openstack-tox-py27/24aedf2/ : FAILURE in 6m 16s\n- openstack-tox-py35 http://logs.openstack.org/21/636821/1/check/openstack-tox-py35/6d578e3/ : FAILURE in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/21/636821/1/check/openstack-tox-py36/d85e304/ : FAILURE in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/21/636821/1/check/openstack-tox-docs/ce6c12d/html/ : SUCCESS in 4m 52s\n- openstack-tox-functional-with-sudo http://logs.openstack.org/21/636821/1/check/openstack-tox-functional-with-sudo/e79dd6b/ : SUCCESS in 6m 21s\n- os-vif-ovs http://logs.openstack.org/21/636821/1/check/os-vif-ovs/3b42b36/ : SUCCESS in 2h 03m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3e55a4a3302ffb94c392139a29f657963ed0f950","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-19 11:50:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2037ea056e8b1f55b9ec90d09899b657f79cd52a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-19 11:52:21.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"da6f50b2643726aafc504a5d2dedb1b92ca20bbd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-02-19 12:22:26.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1f08e363f971fc090058862992017b4c13a289a9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-19 13:25:48.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1a35e78373a26ad177014d8885118c95225e53cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 13:58:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/636821/3/check/openstack-tox-lower-constraints/534a321/ : SUCCESS in 7m 04s\n- openstack-tox-pep8 http://logs.openstack.org/21/636821/3/check/openstack-tox-pep8/1da1065/ : SUCCESS in 7m 01s\n- openstack-tox-py27 http://logs.openstack.org/21/636821/3/check/openstack-tox-py27/afb0928/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/21/636821/3/check/openstack-tox-py35/276eb63/ : SUCCESS in 5m 52s\n- openstack-tox-py36 http://logs.openstack.org/21/636821/3/check/openstack-tox-py36/afa3d74/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/21/636821/3/check/openstack-tox-docs/5b1dace/html/ : SUCCESS in 5m 23s\n- openstack-tox-functional-with-sudo http://logs.openstack.org/21/636821/3/check/openstack-tox-functional-with-sudo/eb11c54/ : SUCCESS in 6m 02s\n- os-vif-ovs http://logs.openstack.org/21/636821/3/check/os-vif-ovs/2483630/ : SUCCESS in 1h 43m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6282279e5bcf68e44b3fc07e36c3db82af35ad9b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-02-19 16:58:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- OVS_HW_offload http://13.74.249.42/21/636821/3/check-os-vif/OVS_HW_offload/c47096e : SUCCESS in 1h 07m 17s (non-voting)\n\nTo re-run the job post \u0027recheck os-vif-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"feecf0f003eb4c3bb8775ef14ff08b815ee0a5d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-02-19 18:25:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef0995b4c4e7dc1be68c2c668d436966a8f210bb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-19 19:03:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dca42ed6b318a9ea5ac54a921cecc1a9137b2134","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-19 22:16:03.000000000","message":"Patch Set 3: Workflow-1\n\nthis is broken i have the fix locally and will push shortly","accounts_in_message":[],"_revision_number":3},{"id":"7b1ff86fbf4795b7f8831a6921edb677f12981c3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-19 22:30:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6241ad2fb3fd8819444a1ddfa1305e905791b11a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-20 02:05:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/636821/4/check/openstack-tox-lower-constraints/435d1de/ : SUCCESS in 7m 18s\n- openstack-tox-pep8 http://logs.openstack.org/21/636821/4/check/openstack-tox-pep8/8aede08/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/21/636821/4/check/openstack-tox-py27/fd11868/ : SUCCESS in 7m 30s\n- openstack-tox-py36 http://logs.openstack.org/21/636821/4/check/openstack-tox-py36/3c7588e/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/21/636821/4/check/openstack-tox-py37/650f6aa/ : SUCCESS in 5m 56s\n- openstack-tox-docs http://logs.openstack.org/21/636821/4/check/openstack-tox-docs/55495ca/html/ : SUCCESS in 7m 15s\n- openstack-tox-functional-with-sudo http://logs.openstack.org/21/636821/4/check/openstack-tox-functional-with-sudo/9a37f3a/ : SUCCESS in 5m 42s\n- os-vif-ovs http://logs.openstack.org/21/636821/4/check/os-vif-ovs/c2c1086/ : SUCCESS in 1h 59m 45s","accounts_in_message":[],"_revision_number":4},{"id":"8a73043116f0be443efb766be559595ae53bfa7a","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2019-02-20 14:47:02.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nNote: minor spellchecks if you want to fix them.","accounts_in_message":[],"_revision_number":4},{"id":"e6b0fa5137548683ce2d9a091fcd5f1a5129fb87","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-02-21 11:40:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"73cd34c626698e48e9d456c9af7163a261f6d18a","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-02-21 11:45:27.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b5fdb1958031f714d046d3db35ba48deae82d660","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-21 11:45:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5b266df503f85c00187d6d618b7b120b637cfbe6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-21 11:51:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"810f5eb257354261d7d5a7a02eca10c57d9475f4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-21 15:16:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"91e1e4b7a68398c3040732737c1b437f7070238c","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2019-02-21 15:55:19.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4bb53635baf9883d65a34a9c87f4bea7cb334141","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 17:26:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/636821/5/check/openstack-tox-lower-constraints/2eb02de/ : SUCCESS in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/21/636821/5/check/openstack-tox-pep8/b64ad24/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/21/636821/5/check/openstack-tox-py27/a5a26f4/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/21/636821/5/check/openstack-tox-py36/a2e66a1/ : SUCCESS in 6m 25s\n- openstack-tox-py37 http://logs.openstack.org/21/636821/5/check/openstack-tox-py37/a861426/ : SUCCESS in 6m 55s\n- openstack-tox-docs http://logs.openstack.org/21/636821/5/check/openstack-tox-docs/c5eccac/html/ : SUCCESS in 5m 58s\n- openstack-tox-functional-with-sudo http://logs.openstack.org/21/636821/5/check/openstack-tox-functional-with-sudo/6b4653d/ : SUCCESS in 7m 47s\n- os-vif-ovs http://logs.openstack.org/21/636821/5/check/os-vif-ovs/a075d9d/ : SUCCESS in 1h 54m 18s","accounts_in_message":[],"_revision_number":5},{"id":"70cce9fe333f88570c169a35add6e9d6b4cadee2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2019-02-25 09:06:25.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2951d57d76e4eb6469c59ea99098a7b30441f1e4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-02-25 09:15:09.000000000","message":"Patch Set 5: Code-Review+2\n\n(5 comments)\n\nNo issues here. Essentially a find and replace","accounts_in_message":[],"_revision_number":5},{"id":"23de27e8c1e178755d0d8577e351ed85da1c2b0a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-02-25 11:15:52.000000000","message":"Patch Set 5: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8890842193caaaf8ffb711aa6e66e803800f30c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-25 11:16:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"67206c539b80a16d21bcc750ba3f55ad2437ac7e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2019-02-25 12:50:42.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\n++ good work, Sean, thank you.","accounts_in_message":[],"_revision_number":5},{"id":"47b69ff8c669763e0d9bf8b068440d3063dedf35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-25 13:02:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/636821/5/gate/openstack-tox-lower-constraints/199f30e/ : SUCCESS in 7m 28s\n- openstack-tox-pep8 http://logs.openstack.org/21/636821/5/gate/openstack-tox-pep8/7311edf/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/21/636821/5/gate/openstack-tox-py27/9652477/ : SUCCESS in 5m 34s\n- openstack-tox-py36 http://logs.openstack.org/21/636821/5/gate/openstack-tox-py36/6286f8d/ : SUCCESS in 6m 36s\n- openstack-tox-py37 http://logs.openstack.org/21/636821/5/gate/openstack-tox-py37/356ec05/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/21/636821/5/gate/openstack-tox-docs/7017855/html/ : SUCCESS in 5m 26s\n- openstack-tox-functional-with-sudo http://logs.openstack.org/21/636821/5/gate/openstack-tox-functional-with-sudo/c196707/ : SUCCESS in 7m 35s\n- os-vif-ovs http://logs.openstack.org/21/636821/5/gate/os-vif-ovs/a34610e/ : SUCCESS in 1h 40m 29s","accounts_in_message":[],"_revision_number":5},{"id":"d02f5a69b37a3269dc40843d63bda111f0eb604f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-25 13:02:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5027ce833c6fccaa80b5ddc8544d262c0bf99dbd","revisions":{"4d2425bfbe0924a721b983f698fd11e21b55fd03":{"kind":"REWORK","_number":1,"created":"2019-02-14 04:07:43.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/21/636821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/21/636821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/21/636821/1"}}},"commit":{"parents":[{"commit":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","subject":"Merge \"Add test to check os_vif.internal.command.ip.exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/192d6fd1149e56bd14f9e0f9fab54c056ab02373"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-14 03:08:53.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-14 04:06:37.000000000","tz":0},"subject":"[WIP] remove brctl from vif_plug_ovs","message":"[WIP] remove brctl from vif_plug_ovs\n\n- This change extends the ip_command interface set function\n  to accept a master as a parent device for a given interface.\n- This change extends the impl_pyroute2 add function to\n  support creating linux bridges.\n- This change replaces calls to brctl with calls to the ip_command api.\n- TODO update unit tests and add fuctional tests.\n\nChange-Id: I8308e8840e20b0a72d00880c1a7996b4c73f6a83\nPartial-Bug: #1801919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/4d2425bfbe0924a721b983f698fd11e21b55fd03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/4d2425bfbe0924a721b983f698fd11e21b55fd03"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9fef3ec2b14fec8d7d3ac7cf42be41a20f86008c":{"kind":"REWORK","_number":2,"created":"2019-02-19 11:50:58.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/21/636821/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/21/636821/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/21/636821/2"}}},"commit":{"parents":[{"commit":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","subject":"Merge \"Add test to check os_vif.internal.command.ip.exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/192d6fd1149e56bd14f9e0f9fab54c056ab02373"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-14 03:08:53.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-19 11:50:55.000000000","tz":0},"subject":"[WIP] remove brctl from vif_plug_ovs","message":"[WIP] remove brctl from vif_plug_ovs\n\n- This change extends the ip_command interface set function\n  to accept a master as a parent device for a given interface.\n- This change extends the impl_pyroute2 add function to\n  support creating linux bridges.\n- This change replaces calls to brctl with calls to the ip_command api.\n\nChange-Id: I8308e8840e20b0a72d00880c1a7996b4c73f6a83\nPartial-Bug: #1801919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/9fef3ec2b14fec8d7d3ac7cf42be41a20f86008c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/9fef3ec2b14fec8d7d3ac7cf42be41a20f86008c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"46edd3849e8400bd9c3facbccbb51bcb73efe619":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-02-19 11:52:21.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/21/636821/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/21/636821/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/21/636821/3"}}},"commit":{"parents":[{"commit":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","subject":"Merge \"Add test to check os_vif.internal.command.ip.exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/192d6fd1149e56bd14f9e0f9fab54c056ab02373"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-14 03:08:53.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-19 11:52:03.000000000","tz":0},"subject":"remove brctl from vif_plug_ovs","message":"remove brctl from vif_plug_ovs\n\n- This change extends the ip_command interface set function\n  to accept a master as a parent device for a given interface.\n- This change extends the impl_pyroute2 add function to\n  support creating linux bridges.\n- This change replaces calls to brctl with calls to the ip_command api.\n\nChange-Id: I8308e8840e20b0a72d00880c1a7996b4c73f6a83\nPartial-Bug: #1801919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/46edd3849e8400bd9c3facbccbb51bcb73efe619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/46edd3849e8400bd9c3facbccbb51bcb73efe619"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6e6e7edcf17ea055978f7360330a45a220e379a2":{"kind":"REWORK","_number":4,"created":"2019-02-19 22:30:11.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/21/636821/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/21/636821/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/21/636821/4"}}},"commit":{"parents":[{"commit":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","subject":"Merge \"Add test to check os_vif.internal.command.ip.exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/192d6fd1149e56bd14f9e0f9fab54c056ab02373"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-14 03:08:53.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-19 22:14:40.000000000","tz":0},"subject":"remove brctl from vif_plug_ovs","message":"remove brctl from vif_plug_ovs\n\n- This change extends the ip_command interface set function\n  to accept a master as a parent device for a given interface.\n- This change extends the impl_pyroute2 add function to\n  support creating linux bridges.\n- This change replaces calls to brctl with calls to the ip_command api.\n\nChange-Id: I8308e8840e20b0a72d00880c1a7996b4c73f6a83\nPartial-Bug: #1801919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/6e6e7edcf17ea055978f7360330a45a220e379a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/6e6e7edcf17ea055978f7360330a45a220e379a2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5027ce833c6fccaa80b5ddc8544d262c0bf99dbd":{"kind":"REWORK","_number":5,"created":"2019-02-21 15:16:54.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/21/636821/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/21/636821/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/21/636821/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/21/636821/5"}}},"commit":{"parents":[{"commit":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","subject":"Merge \"Add test to check os_vif.internal.command.ip.exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/192d6fd1149e56bd14f9e0f9fab54c056ab02373"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-14 03:08:53.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-02-21 14:21:10.000000000","tz":0},"subject":"remove brctl from vif_plug_ovs","message":"remove brctl from vif_plug_ovs\n\n- This change extends the ip_command interface set function\n  to accept a master as a parent device for a given interface.\n- This change extends the impl_pyroute2 add function to\n  support creating linux bridges.\n- This change replaces calls to brctl with calls to the ip_command api.\n- This change removes the use of tee to disable ipv6 in the ovs plugin.\n\nChange-Id: I8308e8840e20b0a72d00880c1a7996b4c73f6a83\nPartial-Bug: #1801919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/5027ce833c6fccaa80b5ddc8544d262c0bf99dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/5027ce833c6fccaa80b5ddc8544d262c0bf99dbd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"192d6fd1149e56bd14f9e0f9fab54c056ab02373","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
