)]}'
{"id":"openstack%2Fos-vif~655068","triplet_id":"openstack%2Fos-vif~master~I2125f9e2912aa7ac5346292d23643fdea003d7c0","project":"openstack/os-vif","branch":"master","topic":"bug/1826826","hashtags":[],"change_id":"I2125f9e2912aa7ac5346292d23643fdea003d7c0","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:40:08.000000000","updated":"2019-05-13 16:48:07.000000000","submitted":"2019-05-13 16:48:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655068-1557766087860-96febb6b","meta_rev_id":"8b835ade5eb2ebdc4cc76bdc055df5eea4bb440a","_number":655068,"virtual_id_number":655068,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-05-13 16:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":1,"date":"2019-04-30 15:22:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-04-30 13:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2019-05-05 17:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-05-01 14:43:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-05-03 18:20:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-13 14:57:51.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-13 14:57:51.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 15:54:14.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-04-30 13:06:28.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-04-30 15:22:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-05-01 14:43:00.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-05-03 18:20:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-05 17:15:46.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-05-13 14:57:51.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-05-13 16:48:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28a99faa7c5bdf53f3d045d9415b97e35d5b3485","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:40:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"901fcac5f6964a6918d1d00a8f5690aba2489116","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:54:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/68/655068/1/check/test-release-openstack/55886df/ : SUCCESS in 4m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/655068/1/check/openstack-tox-lower-constraints/353a3c0/ : SUCCESS in 5m 59s\n- openstack-tox-pep8 http://logs.openstack.org/68/655068/1/check/openstack-tox-pep8/4f8ca18/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/68/655068/1/check/openstack-tox-py27/fd97657/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/68/655068/1/check/openstack-tox-py36/9fb98ca/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/68/655068/1/check/openstack-tox-py37/b72102c/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/68/655068/1/check/openstack-tox-docs/fe69d07/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/68/655068/1/check/build-openstack-releasenotes/ae71fc7/html/ : SUCCESS in 3m 55s\n- openstack-tox-functional-ovs-with-sudo http://logs.openstack.org/68/655068/1/check/openstack-tox-functional-ovs-with-sudo/efdab84/ : SUCCESS in 5m 28s\n- os-vif-ovs finger://ze06.openstack.org/6be9a49d332e474f86361aa5109d88ac : RETRY_LIMIT in 18m 47s\n- os-vif-ovs-iptables finger://ze07.openstack.org/1305a0011c2b4fe5a77f86d4e3ab9976 : RETRY_LIMIT in 12m 51s\n- os-vif-linuxbridge http://logs.openstack.org/68/655068/1/check/os-vif-linuxbridge/288427d/ : RETRY_LIMIT in 21m 13s","accounts_in_message":[],"_revision_number":1},{"id":"97d3adef707febcf2ac61aff7a1a9b426dd4483b","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-04-24 02:00:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6f93e7320003decba18b1119e563fd6c7b658c7","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-04-26 15:54:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62bb528251e9febb6972ab0bf1080d1b9e75ed67","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-29 05:23:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6bbe39cad1bc731485aa88892876f4686c1c072","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-29 05:24:53.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bb6555f440d6615d73240029d72115129767bd84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:52:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/68/655068/3/check/test-release-openstack/ded9ed1/ : SUCCESS in 3m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/655068/3/check/openstack-tox-lower-constraints/2fe7e3e/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/68/655068/3/check/openstack-tox-pep8/3de6b64/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/68/655068/3/check/openstack-tox-py27/dea827b/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/68/655068/3/check/openstack-tox-py36/ff0cdcb/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/68/655068/3/check/openstack-tox-py37/257c762/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/68/655068/3/check/openstack-tox-docs/240faa2/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/68/655068/3/check/build-openstack-releasenotes/da0be73/html/ : SUCCESS in 3m 38s\n- openstack-tox-functional-ovs-with-sudo http://logs.openstack.org/68/655068/3/check/openstack-tox-functional-ovs-with-sudo/953e41a/ : SUCCESS in 4m 33s\n- os-vif-ovs http://logs.openstack.org/68/655068/3/check/os-vif-ovs/95207ce/ : FAILURE in 24m 45s\n- os-vif-ovs-iptables http://logs.openstack.org/68/655068/3/check/os-vif-ovs-iptables/7b7d019/ : SUCCESS in 1h 17m 44s\n- os-vif-linuxbridge http://logs.openstack.org/68/655068/3/check/os-vif-linuxbridge/c35ad11/ : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":3},{"id":"084459cb9db0840c05a742bb8bf91ee79791940e","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-04-30 13:06:28.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a05538f202699017593c7147e85bee74b07492b4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-04-30 14:29:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4d77f48e5a9f4a14c224cef5f8fe55deaf83454d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 15:02:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/68/655068/3/check/test-release-openstack/04ea0ec/ : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/655068/3/check/openstack-tox-lower-constraints/9b0fbc0/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/68/655068/3/check/openstack-tox-pep8/370474e/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/68/655068/3/check/openstack-tox-py27/0dc20e7/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/68/655068/3/check/openstack-tox-py36/90ae6ee/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/68/655068/3/check/openstack-tox-py37/742a059/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/68/655068/3/check/openstack-tox-docs/eb159ae/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/68/655068/3/check/build-openstack-releasenotes/dbb84bd/html/ : SUCCESS in 3m 41s\n- openstack-tox-functional-ovs-with-sudo http://logs.openstack.org/68/655068/3/check/openstack-tox-functional-ovs-with-sudo/0078de3/ : SUCCESS in 4m 47s\n- os-vif-ovs http://logs.openstack.org/68/655068/3/check/os-vif-ovs/66c6321/ : SUCCESS in 1h 49m 57s\n- os-vif-ovs-iptables http://logs.openstack.org/68/655068/3/check/os-vif-ovs-iptables/487fac4/ : SUCCESS in 1h 45m 14s\n- os-vif-linuxbridge http://logs.openstack.org/68/655068/3/check/os-vif-linuxbridge/f0fc4b6/ : SUCCESS in 1h 41m 01s","accounts_in_message":[],"_revision_number":3},{"id":"c3616a782585ad375341187f03f4fec4bbda4be4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-04-30 15:22:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"30dc641b2884c8d7386e001ffc05c84bd439ecd5","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-01 14:43:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f3ad88a2f7ca8ee1cca00832147bf0068844678c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 18:20:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"66bcae0e27b63f689bde1e48817ec5af9e4e7c7e","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-05 17:15:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2796556e93ea24d3e58ee8e60b0a58401b7139c4","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-05-13 14:57:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"27a2e7c2e29e6d6004d7e501508154121ec90708","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 14:58:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d88a88c67b8034b237c468feacaebb8a9abb602","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 16:48:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/68/655068/3/gate/test-release-openstack/82ab22d/ : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/68/655068/3/gate/openstack-tox-lower-constraints/62f5155/ : SUCCESS in 7m 31s\n- openstack-tox-pep8 http://logs.openstack.org/68/655068/3/gate/openstack-tox-pep8/f9be9bf/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/68/655068/3/gate/openstack-tox-py27/a4fb2dd/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/68/655068/3/gate/openstack-tox-py36/4cd2874/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/68/655068/3/gate/openstack-tox-py37/e2e7397/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/68/655068/3/gate/openstack-tox-docs/1fcf67c/html/ : SUCCESS in 4m 47s\n- build-openstack-releasenotes http://logs.openstack.org/68/655068/3/gate/build-openstack-releasenotes/86dbeee/html/ : SUCCESS in 8m 00s\n- openstack-tox-functional-ovs-with-sudo http://logs.openstack.org/68/655068/3/gate/openstack-tox-functional-ovs-with-sudo/e9daf93/ : SUCCESS in 9m 25s\n- os-vif-ovs http://logs.openstack.org/68/655068/3/gate/os-vif-ovs/4cd7924/ : SUCCESS in 1h 48m 04s\n- os-vif-ovs-iptables http://logs.openstack.org/68/655068/3/gate/os-vif-ovs-iptables/5e6a401/ : SUCCESS in 1h 47m 19s\n- os-vif-linuxbridge http://logs.openstack.org/68/655068/3/gate/os-vif-linuxbridge/847a385/ : SUCCESS in 1h 44m 38s","accounts_in_message":[],"_revision_number":3},{"id":"e157b0627757da0366883ec2b3189d776b54e16f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 16:48:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9fb828291c31ca798f454b08ca1288d0e718398f","revisions":{"e9fb11d576836128f043cd13ba1b632316709b12":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:40:08.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/68/655068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/68/655068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/68/655068/1"}}},"commit":{"parents":[{"commit":"1f78150fadbe2796a3d64544dabe28a82ead9865","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/1f78150fadbe2796a3d64544dabe28a82ead9865"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:44.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:44.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I2125f9e2912aa7ac5346292d23643fdea003d7c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/e9fb11d576836128f043cd13ba1b632316709b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/e9fb11d576836128f043cd13ba1b632316709b12"}]},"branch":"refs/heads/master"},"a3e2fc7d8e8be7119bc6cd5d2a0ddb7cd9b04fb0":{"kind":"REWORK","_number":2,"created":"2019-04-29 05:23:51.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/68/655068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/68/655068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/68/655068/2"}}},"commit":{"parents":[{"commit":"1f78150fadbe2796a3d64544dabe28a82ead9865","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/1f78150fadbe2796a3d64544dabe28a82ead9865"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:44.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-29 05:23:38.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I2125f9e2912aa7ac5346292d23643fdea003d7c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/a3e2fc7d8e8be7119bc6cd5d2a0ddb7cd9b04fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/a3e2fc7d8e8be7119bc6cd5d2a0ddb7cd9b04fb0"}]},"branch":"refs/heads/master"},"9fb828291c31ca798f454b08ca1288d0e718398f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-04-29 05:24:53.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/68/655068/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/68/655068/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/68/655068/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/68/655068/3"}}},"commit":{"parents":[{"commit":"1f78150fadbe2796a3d64544dabe28a82ead9865","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/1f78150fadbe2796a3d64544dabe28a82ead9865"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:44.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-29 05:24:37.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I2125f9e2912aa7ac5346292d23643fdea003d7c0\nCloses-Bug: #1826826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/9fb828291c31ca798f454b08ca1288d0e718398f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/9fb828291c31ca798f454b08ca1288d0e718398f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
