)]}'
{"id":"openstack%2Fos-vif~724974","triplet_id":"openstack%2Fos-vif~master~I1b362dba9d46f766202e4329371dca71e2f3b214","project":"openstack/os-vif","branch":"master","topic":"no-babel","hashtags":[],"change_id":"I1b362dba9d46f766202e4329371dca71e2f3b214","subject":"Remove translation sections from setup.cfg","status":"MERGED","created":"2020-05-02 12:11:50.000000000","updated":"2020-05-08 22:08:52.000000000","submitted":"2020-05-08 22:07:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724974-1588975656001-0a89928d","meta_rev_id":"3a52c70db2eaefa8a6d3ea2dc79d972b19240ab2","_number":724974,"virtual_id_number":724974,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-05-08 17:10:22.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 22:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-08 19:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-05-08 17:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-08 19:13:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 17:02:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-08 17:10:22.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-08 19:13:55.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-05-08 22:07:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08316a2d7f20d616bffbfbe958ade8854faa7eba","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2020-05-02 12:11:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cbc9e9f3e93c158639decf9865fea29d3b24272","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-02 12:23:24.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/74/724974/1/OpenVswitch-HW-Offload-os-vif : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0cbfa6ca8aa2638ea848ab0153fb75385689e7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 13:38:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9e79cf916a3e406789287422ab932477 : SUCCESS in 2m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/74a95a080c54404db0f0e673376fd481 : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a249b0a60dc746d5baa7b4a0ff24fd1c : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b335b66d43240c3ada6bdd8792f1127 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c39dfcd8e72c4d598dadbed9990eac9b : SUCCESS in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d036c4c3962b48c8a9c3a76a47c89008 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0662eec81d04c8e9b69a542a5e3f559 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f64ad7eafc4e1c8ab904d621f46c35 : SUCCESS in 4m 11s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/689abe40ba994022a9f666fa83d56fc5 : SUCCESS in 1h 17m 43s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/d2f30c5b4b0f4492aecf4b34a4b217b9 : SUCCESS in 4m 37s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/b350d1e736f945fea449e1d1ddf29e96 : SUCCESS in 1h 16m 23s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/f7f3e511af6b4bfd9af94aadf21cbe41 : SUCCESS in 1h 20m 10s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/f1721ce1c457434e8ca41eefc2579752 : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b9706c2bbe990b02090f16d10a9de1d69945b725","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-07 09:45:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d25791c815882562098545996ae2a3852bae8f3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-05-07 14:38:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nbable isnt a requirement directly but its pulled in by \noslo.i18n and there for its in the lower constaits since its an indirect depency\nhttps://github.com/openstack/os-vif/blob/master/requirements.txt#L10\n\nso if you want to remove bable you need to remove oslo.i18n\nas it stands this patch is incorrect as it will uncap bable in the lower constratis job but it will still be installed.","accounts_in_message":[],"_revision_number":1},{"id":"048172706f28d5456a19d8d8dbd1cbac6182a7df","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2020-05-08 16:59:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a416c352ae8cdee713cf0aa8fb61bcd4b3785037","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-08 17:02:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d0672e218476d2b67cbd1caf5cc45d76ba7967c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-05-08 17:10:22.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/74/724974/2/OpenVswitch-HW-Offload-os-vif : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"667f25f17bd247cb2946c8bc1ea6d1030869e952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 18:39:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2924768aaca455fafdbb8c4fbdba8a2 : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af41f40855a544758fd7a1b61479527c : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f505e45cada84f9da408472cbb803643 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2075dd7f797a40a2962eeadcc84550ba : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/454d4b5e3efd4d619a8149f407008d72 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e6f79a35d5d49f38f1540be51dde8d8 : SUCCESS in 4m 06s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/75c11af44e534ce9a6e1b0af8e410b54 : SUCCESS in 1h 16m 23s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/25b1335ed9174d99a1f527997b1a437b : SUCCESS in 5m 02s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/458b4c27a45942ceb786d8a02c3b57dd : SUCCESS in 1h 34m 01s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/878ca719f09a4bdd9a1ccf6c16570869 : SUCCESS in 1h 12m 52s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/39a633a7301246bb84b1d722c793a284 : SUCCESS in 1h 37m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4e1a3d4a77605791c47850e5926b52dbb5caa1db","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-05-08 19:13:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nyep im good with this.\n\nas a libviary os-vif was never translated anyway so this cam form the cookiecutter template i think orgininally so its good to drop it","accounts_in_message":[],"_revision_number":2},{"id":"8e1f833a65ebb2f2a1088b206f65b54b5028ae60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 19:14:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e400ed53cd05b3a526f350cc4d1f901e6b2c5f5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 20:18:29.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b235038f6995c4d0d13998eddb7b7da155449011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 22:07:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a4b165fe72d44af3850dea44dff9950b : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d190f75d13d743eeb07e112e68323649 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4eb4c78cf5f404f9e377674d9a2baf0 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd13aabd34f4442b9a6e06eece01cc34 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65218c86d4e9475e87326f195c354a4a : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080a2caf27cb42268b1ce75bf8fbe5ab : SUCCESS in 5m 39s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/c682c04dd0fa41279fe43ec887a6272d : SUCCESS in 6m 49s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/be50d5814864421eb3fb555f9d7361f8 : SUCCESS in 1h 25m 08s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/ff41031cff1f4d6f9bbb59e414d23abe : SUCCESS in 1h 32m 59s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/c6308c94de844f59a85d78098a5c429a : SUCCESS in 1h 40m 26s","accounts_in_message":[],"_revision_number":2},{"id":"f4b945a0219937d727486918a62b28d24dae307f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 22:07:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b2013283c4e5ad30cb3ebb478834b6f67a0b4d14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 22:08:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddbba234d994478e889dde4e0bc70522 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"64013a860986d0b5b9a37f440f19286f1948ea74","revisions":{"9c580cff382f7d19eef1686023d44ae95b47cae0":{"kind":"REWORK","_number":1,"created":"2020-05-02 12:11:50.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/724974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/74/724974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/74/724974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/74/724974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/74/724974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/74/724974/1"}}},"commit":{"parents":[{"commit":"222e2866615e65ece5a1575326e082bb75c5b544","subject":"Merge \"trivial: Remove some rules from flake8 ignore list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/222e2866615e65ece5a1575326e082bb75c5b544"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2020-05-02 12:06:33.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2020-05-02 12:10:45.000000000","tz":540},"subject":"Remove Babel requirement","message":"Remove Babel requirement\n\nBabel is not needed as requirement, remove it.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014227.html\n\nChange-Id: I1b362dba9d46f766202e4329371dca71e2f3b214\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/9c580cff382f7d19eef1686023d44ae95b47cae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/9c580cff382f7d19eef1686023d44ae95b47cae0"}]},"branch":"refs/heads/master"},"64013a860986d0b5b9a37f440f19286f1948ea74":{"kind":"REWORK","_number":2,"created":"2020-05-08 16:59:50.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/724974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/74/724974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/74/724974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/74/724974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/74/724974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/74/724974/2"}}},"commit":{"parents":[{"commit":"1aa4da7c7a2e520e52127cb4afa590e9ed1ba2e3","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/1aa4da7c7a2e520e52127cb4afa590e9ed1ba2e3"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2020-05-02 12:06:33.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2020-05-08 16:58:42.000000000","tz":540},"subject":"Remove translation sections from setup.cfg","message":"Remove translation sections from setup.cfg\n\nThese translation sections are not needed anymore,\nBabel can generate translation files without them.\n\nRemove babel.cfg as well, this is the default role and not needed\nanymore.\n\nChange-Id: I1b362dba9d46f766202e4329371dca71e2f3b214\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/64013a860986d0b5b9a37f440f19286f1948ea74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/64013a860986d0b5b9a37f440f19286f1948ea74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
