)]}'
{"id":"openstack%2Fos-vif~745895","triplet_id":"openstack%2Fos-vif~master~I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7","project":"openstack/os-vif","branch":"master","topic":"switch-dev-link","hashtags":[],"change_id":"I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7","subject":"Fix - os-vif fails to get the correct UpLink Representor","status":"MERGED","created":"2020-08-12 13:54:17.000000000","updated":"2020-10-22 04:50:12.000000000","submitted":"2020-10-22 04:49:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":71,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745895-1603342148941-c5aa70a3","meta_rev_id":"590aa291a87c5313eb18b5bf6f6d031ca149f3b4","_number":745895,"virtual_id_number":745895,"owner":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-10-22 01:41:19.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 04:49:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-21 11:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2020-10-21 09:50:34.000000000","_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":2,"date":"2020-10-21 09:59:34.000000000","_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":1,"date":"2020-10-21 09:59:34.000000000","_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"value":0,"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 13:54:28.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-10-15 07:32:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-10-18 13:06:20.000000000","updated_by":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"},{"updated":"2020-10-21 09:50:34.000000000","updated_by":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"reviewer":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"state":"REVIEWER"},{"updated":"2020-10-21 09:59:34.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2020-10-21 11:48:57.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-10-22 01:41:19.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-22 04:49:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05cd5d3c518eed23d1a74ad27b113bb918af5795","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-08-12 13:54:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5ff17b1b8dd381029b94149ec5a17e926d64c07","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-08-12 13:54:28.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"5536a753473c8c5b002c4842fbaba1d89b5f77e0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-12 13:58:09.000000000","message":"Patch Set 1:\n\nill try and review this properly later this week\nbut can you confirm if this will work with both the old kernel and new kernel.\n\nwe need to support both in os-vif","accounts_in_message":[],"_revision_number":1},{"id":"8c24582fb801305ec9f5c0cf495a654970678735","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-12 15:11:28.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/1/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f4bba8243ee07f16d6c524ce86721aa5ae44df70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 16:01:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cdc473e71c54b57bf9e5923f497012f : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dcf1bb396bd4eb28f77bbd842914b1b : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c500ea085a940b9b4040f593ff51653 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1339144018b43f59d50728fedf65084 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6675be938ad645d5aca749397052720f : SUCCESS in 3m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/814630409bab4eed8ab0eb25249c752d : SUCCESS in 1h 09m 35s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/7d9d76ce6b7a44fc95e0805791208424 : SUCCESS in 6m 39s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/9cbbebf3993e4535a067515a72860dd1 : SUCCESS in 1h 36m 13s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/f9312a5e9ef74103a5433a81efb0de60 : SUCCESS in 1h 17m 16s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/5c25130e81a04c01acc827a2dcf948ab : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5f96706584fccf5db4669412677b46ce1da95225","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-08-13 12:23:37.000000000","message":"Patch Set 1:\n\nIt is working on the both kernels","accounts_in_message":[],"_revision_number":1},{"id":"63adafbfb9a9cf9a2d37864cd45320c8547b9852","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-13 14:04:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c42e1e4837a0815a9ebc1c9d7c62ec52727a254f","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-08-13 16:19:40.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI will give this a deeper look early next week.","accounts_in_message":[],"_revision_number":1},{"id":"eab4e7bb7527833411816e08d0505a91560bf59c","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-08-17 11:42:32.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ef3ca2ff9082d6e9c930102002229b4d32bbebe0","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-08-17 11:53:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b3bad15d79130106056b33be4a0d70225e29dfe","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-08-17 11:55:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f1afa466c1444481210999bd1b2a3455634cb2f","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-08-18 11:13:32.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThere now 2 patches one for refactoring 746673 and this is for the fixing, this patch is depending on the refactoring one","accounts_in_message":[],"_revision_number":1},{"id":"60bab111203ac89955a7eae2c2920654cf079651","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-08-19 06:24:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24881bccd6c0d66bc1940ca4564ce129c17056ba","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-08-19 06:25:29.000000000","message":"Patch Set 2:\n\n(18 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c380f64410b6a14d5377268da0d15c8ce05cfded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 08:07:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1331089bd8f4f6f91afeeac60f79f7c : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a118220c7c646539725885aa5f15611 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b183a49eee4ff28f17e4c315cf1cfb : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/897733d5f3f346eb95040f05efaa2153 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbbd36933c3e465692e651c26b37e6be : SUCCESS in 4m 12s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ec3c9ddb2f2c466a872bf05b2f326df6 : FAILURE in 1h 23m 57s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/9d3b64f0ddea4fc29dee4cf482f361f0 : SUCCESS in 4m 57s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/1f19b59d072242edba28f6a14181f1af : SUCCESS in 56m 31s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/8a69459d772e4ca5b4fa7ce51005c9fa : SUCCESS in 1h 14m 35s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/a8278108392844569cf9232a8957a3e4 : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":2},{"id":"10d79fa01e32cb162fc3c37f1a6e4aea093c6e06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-08-19 08:49:06.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/2/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4bf7a8af1b8c5e9f9e5d9f91bce26863c4051017","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-08-19 13:00:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8a2c89c0a90c43a6a227f7ec31c351a7452bdc6","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-08-19 13:28:53.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"6a952791f8ceffc4a63e5d57f837712a6b01517c","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-08-19 14:29:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7d00e87608f9ee4411c59d5d40d34f382541767","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-25 16:28:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"872e2970861a6783ab4d3d865fb95243593fea1c","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-09-06 07:39:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f173555cd77fcdbbbc7b1276a06ff192f5c5f373","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-06 08:49:07.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/3/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"bb309135ed33ef13afa27cffd645305123fa6b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 08:51:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df561af8b22046da9c1705e273eda893 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf21e12360b44ab8b449defa5fe01183 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/396f59742fb44f128507d3f9fd47d729 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbddad6313494ddf99d2206cefc9f1b2 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51282e4ce8334e80a649bcf6b3c4d5fb : SUCCESS in 9m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5de8411b563445bda0d920f29d650b09 : SUCCESS in 5m 30s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8f6ebbf7565b4af6863a5affbf2f7f60 : SUCCESS in 57m 58s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/4d57a67fd3ed485d8742b6342904b264 : SUCCESS in 5m 29s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/c8e2cfecd0224eb6a0cc7e0c85db3dbf : SUCCESS in 51m 20s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/6a3b1b9c90c94e87bc9b4b9b673e623b : SUCCESS in 1h 08m 08s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/83ad3a37aa514e1aaf23a2233d6482fa : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d1848742842a1dc6b64c64cddb3f0fd987882af6","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-09-08 14:35:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a66fbba30c9bdfe5e6f55b9f98b747df84728ff4","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-09-08 15:07:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c8504f754b57d2ed36f4701ebcbfd5626910c5d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-09-08 15:29:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f315a1ea0202bd2344f5454c5cacc9ef2418780","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-09-20 08:42:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"73de882f5081b02fb5086471414fcd78905dfda4","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-09-20 08:45:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30aaa2364341065c3ff7e6ab1817aa6fa7723691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 09:57:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b43431b0e4a4dfba5b0038be090bb2f : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f88535f336004633b0343c90ffecfb3a : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5674263b36c84d18b97bb39bf38e475e : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82d62dbef16c4045b2c52946d1275426 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc3a775cfc34e3883c8d50a6dccb484 : SUCCESS in 11m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd6928273bc24cf3b497e01b4cd5ca4d : SUCCESS in 3m 38s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d558f2fb1c4c4a7280b8a811323578b3 : SUCCESS in 1h 02m 33s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/79be7566dac54eda849dc5c08e74b2a2 : SUCCESS in 4m 05s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/5ab8c722d44a4ce09e17cb726b6d3fe1 : SUCCESS in 1h 12m 31s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/38dd092b4b3a4c9690abd92285b089fa : SUCCESS in 1h 13m 25s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/2ddc1daf34b240fd87fbc78549f793f6 : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":4},{"id":"025f7a9fdd120b12006a1a24cae7ce8c504011de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-20 10:02:05.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/4/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"62e02f63f631884a9611e08e190b6e9ba3c40b01","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-09-22 08:48:12.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8667e9791dbf85a283f04fde1366f03856b1e018","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-09-22 12:29:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1b1bcef272266186ed59d2909d949830f58a9f1f","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-09-22 12:30:33.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cab73cbfee111ed7af007213651515a852df56d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 13:53:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a58008dd005b4962a4cbfc346ad1c78a : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a739eebf7b4b45f6bbfd23e03ba9dcae : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4933379224a7440880f98d542d29bdeb : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/317710a3930f4739a5b672c8da54d909 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dec9aa1b99c441ba905138e00f2e1fa4 : SUCCESS in 15m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7ef4d0be09740a3a704928c5ffc19ef : SUCCESS in 4m 32s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1df4affd240a48399c1ff5ed1f6ba3e2 : SUCCESS in 1h 03m 29s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/a9a3d78c7e4a47b293de57fde6164e7a : SUCCESS in 4m 42s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/638936b57479496fa34869e93ddcc768 : SUCCESS in 56m 23s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/977ac0227d36415e849d1f38b1251f35 : SUCCESS in 1h 14m 04s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/103e9ef13fe5456eb17b5781ece24877 : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":5},{"id":"42dbe8c6b3eb34bec10f0d74c1894c59a5bba13f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-09-22 14:10:23.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/5/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"9d48fcc3584505226aa7c7f724a192135121e2b5","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-10-04 12:55:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1a57aee14f5f50bfea4792707b3996b3d60036e0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-05 13:50:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc0fd56220320eb3c919dc27090e2feed1f5d84b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-05 13:58:46.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"395374dca38e3d70e8e075ceeb79955f51d3b5f3","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2020-10-05 14:43:26.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nJust a quick reply, not a full review, looks good so far though.","accounts_in_message":[],"_revision_number":5},{"id":"144881f559164b50c5dcabece30849423ef20af3","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2020-10-05 21:01:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"17448a96e0897321d1bbf2df5cce8b7dc11cdf71","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-10-14 12:52:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c1b9d20d279cb2e80f695690cde5a31bad318302","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-10-14 12:55:02.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"994a9c86072ecd852ab2328ef1f6c3ff5863d858","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-14 14:08:50.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/6/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"d784adf38cf789fc3d592dd3a81a1f822905bcc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 14:32:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a907c04563944b4b88360f47de74c264 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c0d8a512174ebebd88e6266179e0f2 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba17104cca754b658644a8f134cceaf8 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41cbe4a8d5b7424a9d252982131cd695 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16700365b80c42799cf18bc79a2240e2 : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a154e4e52584dffbf5849c7d3565b42 : SUCCESS in 3m 51s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f09c5a33109240308696a683e18b9e72 : SUCCESS in 1h 08m 55s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/d035e311ef3343d19fa212aae1b4aacd : SUCCESS in 4m 54s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/d2780f1f8aaa4ec486dcb05f0ff0bde2 : SUCCESS in 1h 19m 16s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/132bc6ddd2964d8caa6f7427199462e4 : SUCCESS in 1h 03m 33s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/ec95973dcd1d40b59f7f3847001714f4 : SUCCESS in 1h 29m 04s","accounts_in_message":[],"_revision_number":6},{"id":"3d5f760dfc9a077fc85425055793832a3ffae792","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2020-10-14 18:20:11.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f48257b6e652622dff63d0f2aa89d2f232d8ddd3","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-10-18 13:06:20.000000000","message":"Patch Set 6: Code-Review+1\n\nThis seems OK to me, main difference between PS5 -\u003e PS6 is \nclarification of unit-test and keeping original behaviour of using the first switchdev netdev encountered in case no matching uplink representor found.","accounts_in_message":[],"_revision_number":6},{"id":"b2c55ff6b5c82b7ca117ab293a9b19348e964e86","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-10-18 13:16:07.000000000","message":"Patch Set 6:\n\nAlso i believe Jan\u0027s comment about HW VEB refers to:\n\nhttps://github.com/openstack/nova/blob/20ca7c4337adb58d9290521fb20803d789f2b574/nova/pci/utils.py#L148","accounts_in_message":[],"_revision_number":6},{"id":"6541d392937f3e405e72e4d102613d6506f65679","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-19 12:06:48.000000000","message":"Patch Set 6:\n\nim going to ask jan to take one more look at this while i review it.\nbut i think we can proceed with merging this","accounts_in_message":[],"_revision_number":6},{"id":"3cc033400ad3297bae544e79bf65b5c2ca3cf475","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-19 12:24:50.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)\n\npendign jans review im +1\ni would still like to see a release note and if jan is ok with this i think we can merge it today and start the backport.","accounts_in_message":[],"_revision_number":6},{"id":"4e316e39a5613e39d9696c0ab4fff9372774e85f","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2020-10-19 16:32:33.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)\n\nFrom my side, it looks good:\n* Thanks for cleaning up the tests, it\u0027s a much needed improvement in readability\n* Would you consider adding the following release note?\n\n---\nfixes:\n  - |\n    Linux kernel 5.8 changed the sysfs interface that is used to\n    discover the interfaces used for OVS offloads for certain NIC\n    models. This results in network plugging failure, as described\n    in `bug #1892132`_. This release fixes the plugging issue by\n    properly handling the new sysfs structure.\n\n    .. _bug #1892132: https://bugs.launchpad.net/os-vif/+bug/1892132","accounts_in_message":[],"_revision_number":6},{"id":"cfcacafbb19fe0bce63ceb94b9977421012dc490","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-19 16:42:01.000000000","message":"Patch Set 6:\n\nthanks jan.\n\nok if you add the release note im happy with fast approving the new versions","accounts_in_message":[],"_revision_number":6},{"id":"cd5c48189541a4ab6a0f50cbf01505867e5d995b","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2020-10-19 16:42:05.000000000","message":"Patch Set 6:\n\nPlease consider my +1 a +2 (for fast approval) if an acceptable release note is added!","accounts_in_message":[],"_revision_number":6},{"id":"f8b8e63e15d1bec28a052072d3229ba3a1a5e90c","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2020-10-19 18:31:43.000000000","message":"Patch Set 6: -Code-Review\n\nNeed to add release note as Jan mentioned","accounts_in_message":[],"_revision_number":6},{"id":"d8dce95fc42e63a73dc5251911ff331207afc63c","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-10-21 07:03:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fddadda89dcb68a0544041a079b09bf6e07a8ddc","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-10-21 07:35:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a0109b6292f22dfeea51d72330e7ec9fcc9f3866","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-21 07:36:18.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/7/OpenVswitch-HW-Offload-os-vif : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"ad3c339e5c13c5fe5ff73f38c53a42cda48e61b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-21 08:53:01.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/8/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"c20a644abb1618775454eec20b3244712e941fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 09:19:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e1989669a24268997b4d49cd3cbb9a : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7144d99d81ae470ca21d4de385956659 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1fc481acce74caca277bec1c7276b53 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1372a48b0bf24af4b4cf38597793de83 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc09506990fd468ea56ec50bcdc6bdcc : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed3d07a57fb463fb011b332050d831d : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a01255bb08904d6c8038aad6207d535f : SUCCESS in 5m 37s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5c3b6c849a394869a40ca52d804ed37f : SUCCESS in 1h 10m 35s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/b032a7be60c94c6e925e6dae1a3c0890 : SUCCESS in 5m 31s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/39e8c21b4d814c06ab7f12d66bf1b0e0 : SUCCESS in 1h 23m 11s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/771e9bb083484cf5bd7ca89572dd0229 : SUCCESS in 1h 07m 32s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/487fb82261564a2cad8bfceb1a9ad0b5 : SUCCESS in 1h 29m 14s","accounts_in_message":[],"_revision_number":8},{"id":"40edc01e747ce8e248a1f446072ae3cc568dcd91","author":{"_account_id":25733,"name":"Jan Gutter","email":"github@jangutter.com","username":"jangutter"},"date":"2020-10-21 09:50:34.000000000","message":"Patch Set 8: Code-Review+2\n\nThank you very much for your patience and contribution!","accounts_in_message":[],"_revision_number":8},{"id":"fde9ac4ef572d152e942dff296978acd2012f6f4","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2020-10-21 09:59:34.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"36ac23c51cd55530a1e53e067cb582e511d3f17f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 09:59:45.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c42397100c1de9d5ce12307eac068e7ea1006de4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-21 11:48:57.000000000","message":"Patch Set 8: Code-Review+2\n\nthanks for the release note","accounts_in_message":[],"_revision_number":8},{"id":"557200335cf2e1fbffe7a0fc4ce4327b925c6dee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 12:03:04.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba562d4a719141a9979b04ad582a06e5 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05686269b63f461e9a93377f227d4c94 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7daeab541214d5789b9610c224a4ae0 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/747cacacc41e4f7a99e4fb6f878859fc : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c901613573b480687d569005d6d23ed : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dfbd9cf7d144bfaaf24636c87074335 : SUCCESS in 6m 23s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/c3e86f3d6dc84799ab3348b60c9b01da : SUCCESS in 6m 57s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/083f20a48535447494120522c4a5a72f : SUCCESS in 1h 21m 31s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/d01bced0e20f4bcca40237bb21093661 : FAILURE in 7m 29s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/7e72c64bccef4c5fb545fe75f6bf422d : SUCCESS in 1h 29m 49s","accounts_in_message":[],"_revision_number":8},{"id":"e8222dfc94fdfb6b4e64968e12ff900348027908","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-21 12:29:49.000000000","message":"Patch Set 8:\n\nthe iptables job failed because of a missing -6 option in iproute","accounts_in_message":[],"_revision_number":8},{"id":"1e41cadcd3c5af9ad851385b50bcb2bd9b29effa","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-21 12:30:22.000000000","message":"Patch Set 8:\n\ncaused by https://github.com/openstack/devstack/commit/0545b48f3b14f2951033b5e09db3190a95cf3527","accounts_in_message":[],"_revision_number":8},{"id":"21825ce284a934c7d12a6b6f43b4113108606617","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-10-21 18:14:50.000000000","message":"Patch Set 8:\n\nrecheck this seams to be an issue with the limestone provider","accounts_in_message":[],"_revision_number":8},{"id":"39e7242514a79164d7e50233d7b9b10c0f2f877e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-21 19:30:57.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/8/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"bb02b38eecbaf42fd8c60c7da450005baf84d883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 19:57:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b051fe4817b48fea8ec0477b73a9962 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6930b0a4a4c64569b7ca6a266ba58c48 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2887c0f71cf49c5820ae9aa6d1fe80f : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8094c205a6394761ad894f131053e780 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf0a74f81e149fc920d38855167fb17 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c92eaeb8a35341878af2c979db0ab91f : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de350e19b5524d8296b29840b33f7b29 : SUCCESS in 6m 23s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d94013f0fa604c8191cba7cabba4347b : SUCCESS in 59m 42s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/c3e9c3ef4cf04147bec1637ef397a9ca : SUCCESS in 6m 13s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/f2bb37b451994581873be8cce0927041 : SUCCESS in 1h 25m 55s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/01286e1d4ac846d7828043be4579ca6a : SUCCESS in 1h 35m 04s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/6cb46dc9bea44b318ebb39cee10db811 : FAILURE in 1h 11m 44s","accounts_in_message":[],"_revision_number":8},{"id":"8ce6c1f1bcd4ba5232dc1813a73d3b62a3bc2312","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2020-10-22 00:30:57.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"08668659f0a0f205b694d1911f6fb7f36131b37d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-10-22 01:41:19.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/95/745895/8/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"e748aca8ed7888adda4348af8ae6851e5945c3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 02:36:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/074b356f6dac4fecb519c05f00dfa1e7 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a467600b544f268466a5698014ade8 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f0271647462492f877ee9a180379377 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5785a85a17bb4f04a602e8b8624dbc55 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5fbbb206b34315ad85c1915a7aa65a : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e08e7d2efb484153853a5f466e088b5c : SUCCESS in 4m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/540984aba1534bec858c491d9045a9bc : SUCCESS in 5m 12s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/21d519398fdb4fda9d17eb29dcec3f2f : SUCCESS in 57m 46s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/f84790d085a54fbfb8fe69c1104e241e : SUCCESS in 5m 12s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/24397cf581dd4abb82ca6a004095819a : SUCCESS in 1h 16m 07s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/1c1349565f3e4f6d956460d25edacd70 : SUCCESS in 1h 28m 43s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/92cb84c76c2141728df68e82333e01ef : SUCCESS in 1h 27m 10s","accounts_in_message":[],"_revision_number":8},{"id":"60df5b2c45203dcbc4ee96d1aaca21f40c820995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 02:36:31.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"502c431b0a585ae886cae99ef19ee7fa58c54e90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 04:49:08.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c133c21dc3f14065b0f25020aed7731d : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26429a98a7434638b1224a55f6174ddd : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d0bbe055f22400d8f47505dec67bf7e : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9037dd822fe4b56b9a42572cd04fdb7 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd827769513b4c6b9910a884b9334b5a : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1179b39737364204a46c296df1f4d85f : SUCCESS in 6m 19s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/2a5fca90f9ae4227b493245710d5cc9d : SUCCESS in 7m 02s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/bec1f3dbaf1d4e9389279b933635d8ba : SUCCESS in 1h 27m 13s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/4d5b1644edfc4cef99920c3cdbf18325 : SUCCESS in 1h 26m 45s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/f1071185beee42df90a33d2fca10b9d9 : SUCCESS in 1h 32m 18s","accounts_in_message":[],"_revision_number":8},{"id":"a4db3903c03eeb0876008eefbab2c24dbd886375","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 04:49:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"4150b2a96fec6822f04e5757bc7440cdedcfe900","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 04:50:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be942a0998a447a91f08d94d3aaeae7 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45a73e4d0cea47e6a9f7cc8552c8acf5 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b37de19c58c877f5174d76d0a4ba5ab519f464e8","revisions":{"e4a7101f9ea811f07a6e45bcd045dc429108e931":{"kind":"REWORK","_number":1,"created":"2020-08-12 13:54:17.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/95/745895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/95/745895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/95/745895/1"}}},"commit":{"parents":[{"commit":"041ce67619074ee79226161aab8169f990e4c175","subject":"[goal] migrate testing to ubuntu focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/041ce67619074ee79226161aab8169f990e4c175"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-08-12 13:54:14.000000000","tz":180},"subject":"Fix - os-vif fails to get the correct VF representor and UpLink Representor","message":"Fix - os-vif fails to get the correct VF representor and UpLink Representor\n\nDue to the new change in kerenl [1], the PF and VF representors are linked\nto their parent PCI device. This patch fixes the failures caused by the new\nkernel change.\n\n[1] https://git.kernel.org/pub/scm/linux/kernel/git/netdev/net.git/commit/?id\u003d123f0f53dd64b67e34142485fe866a8a581f12f1\n\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/e4a7101f9ea811f07a6e45bcd045dc429108e931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/e4a7101f9ea811f07a6e45bcd045dc429108e931"}]},"branch":"refs/heads/master"},"841e4e1c42dfdcc8e8eee6a25c3ced706b57bebd":{"kind":"REWORK","_number":2,"created":"2020-08-19 06:24:59.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/95/745895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/95/745895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/95/745895/2"}}},"commit":{"parents":[{"commit":"aa08372c7407e97e44775b831be40df93dce6019","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/aa08372c7407e97e44775b831be40df93dce6019"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-08-19 06:23:57.000000000","tz":180},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nDue to the new change in kerenl, the PF and VF representors are linked\nto their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/841e4e1c42dfdcc8e8eee6a25c3ced706b57bebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/841e4e1c42dfdcc8e8eee6a25c3ced706b57bebd"}]},"branch":"refs/heads/master"},"4f5c0e5a013e502b13775f1b97438f7a1288e292":{"kind":"REWORK","_number":3,"created":"2020-09-06 07:39:16.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/95/745895/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/95/745895/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/95/745895/3"}}},"commit":{"parents":[{"commit":"76f7565b99e637d74878955a0033f35e9eb0e13f","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/76f7565b99e637d74878955a0033f35e9eb0e13f"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-09-06 07:37:59.000000000","tz":180},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nDue to the new change in kerenl, the PF and VF representors are linked\nto their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/4f5c0e5a013e502b13775f1b97438f7a1288e292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/4f5c0e5a013e502b13775f1b97438f7a1288e292"}]},"branch":"refs/heads/master"},"71177859540de35ca7eeec24b11491a8c8df1d99":{"kind":"REWORK","_number":4,"created":"2020-09-20 08:42:43.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/95/745895/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/95/745895/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/95/745895/4"}}},"commit":{"parents":[{"commit":"76f7565b99e637d74878955a0033f35e9eb0e13f","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/76f7565b99e637d74878955a0033f35e9eb0e13f"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-09-20 08:41:46.000000000","tz":180},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nDue to the new change in kerenl, the PF and VF representors are linked\nto their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/71177859540de35ca7eeec24b11491a8c8df1d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/71177859540de35ca7eeec24b11491a8c8df1d99"}]},"branch":"refs/heads/master"},"664f8931afc564381e71b6c3c996b39d26e82941":{"kind":"REWORK","_number":5,"created":"2020-09-22 12:29:28.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/95/745895/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/95/745895/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/95/745895/5"}}},"commit":{"parents":[{"commit":"76f7565b99e637d74878955a0033f35e9eb0e13f","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/76f7565b99e637d74878955a0033f35e9eb0e13f"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-09-22 12:29:04.000000000","tz":180},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nDue to the new change in kerenl, the PF and VF representors are linked\nto their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/664f8931afc564381e71b6c3c996b39d26e82941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/664f8931afc564381e71b6c3c996b39d26e82941"}]},"branch":"refs/heads/master"},"3b51acafec0ae970bcb61f282fa9827379b008df":{"kind":"REWORK","_number":6,"created":"2020-10-14 12:52:02.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/95/745895/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/95/745895/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/95/745895/6"}}},"commit":{"parents":[{"commit":"76f7565b99e637d74878955a0033f35e9eb0e13f","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/76f7565b99e637d74878955a0033f35e9eb0e13f"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-10-14 08:55:57.000000000","tz":180},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nTill kernel 5.7 PF and VF representors are exposed as virtual device.\nThey are not linked to its parent PCI device like how uplink\nrepresentor is linked.\n\nStarting from kernel 5.8 due to new change [1] the PF and VF representors are\nlinked to their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\n[1] https://git.kernel.org/pub/scm/linux/kernel/git/netdev/net.git/commit/?id\u003d123f0f53dd64b67e34142485fe866a8a581f12f1\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/3b51acafec0ae970bcb61f282fa9827379b008df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/3b51acafec0ae970bcb61f282fa9827379b008df"}]},"branch":"refs/heads/master"},"eb8888796f2a30bcf9e6af11795381cbd826bb7f":{"kind":"REWORK","_number":7,"created":"2020-10-21 07:03:07.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/95/745895/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/95/745895/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/95/745895/7"}}},"commit":{"parents":[{"commit":"76f7565b99e637d74878955a0033f35e9eb0e13f","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/76f7565b99e637d74878955a0033f35e9eb0e13f"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-10-21 07:02:44.000000000","tz":180},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nTill kernel 5.7 PF and VF representors are exposed as virtual device.\nThey are not linked to its parent PCI device like how uplink\nrepresentor is linked.\n\nStarting from kernel 5.8 due to new change [1] the PF and VF representors are\nlinked to their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\n[1] https://git.kernel.org/pub/scm/linux/kernel/git/netdev/net.git/commit/?id\u003d123f0f53dd64b67e34142485fe866a8a581f12f1\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/eb8888796f2a30bcf9e6af11795381cbd826bb7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/eb8888796f2a30bcf9e6af11795381cbd826bb7f"}]},"branch":"refs/heads/master"},"b37de19c58c877f5174d76d0a4ba5ab519f464e8":{"kind":"REWORK","_number":8,"created":"2020-10-21 07:35:37.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/95/745895/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/95/745895/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/95/745895/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/95/745895/8"}}},"commit":{"parents":[{"commit":"76f7565b99e637d74878955a0033f35e9eb0e13f","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/76f7565b99e637d74878955a0033f35e9eb0e13f"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-10-21 07:35:22.000000000","tz":180},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nTill kernel 5.7 PF and VF representors are exposed as virtual device.\nThey are not linked to its parent PCI device like how uplink\nrepresentor is linked.\n\nStarting from kernel 5.8 due to new change [1] the PF and VF representors are\nlinked to their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\n[1] https://git.kernel.org/pub/scm/linux/kernel/git/netdev/net.git/commit/?id\u003d123f0f53dd64b67e34142485fe866a8a581f12f1\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/b37de19c58c877f5174d76d0a4ba5ab519f464e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/b37de19c58c877f5174d76d0a4ba5ab519f464e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
