)]}'
{"id":"openstack%2Fos-vif~765970","triplet_id":"openstack%2Fos-vif~stable%2Ftrain~I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7","project":"openstack/os-vif","branch":"stable/train","topic":"train-fix-swirchdev","hashtags":[],"change_id":"I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7","subject":"Fix - os-vif fails to get the correct UpLink Representor","status":"ABANDONED","created":"2020-12-08 11:24:53.000000000","updated":"2023-09-01 17:07:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"85a6c862e6b3b603f04496266c241163702b3467","_number":765970,"virtual_id_number":765970,"owner":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-05 09:05:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2022-02-08 16:48:40.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 11:26:21.000000000","updated_by":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-12-08 11:26:21.000000000","updated_by":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"},{"updated":"2020-12-08 11:26:21.000000000","updated_by":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2020-12-08 11:54:44.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-08 11:56:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-03 16:50:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2021-08-19 08:48:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-09-08 11:25:57.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"d92fe2478fae857b5e1a2a09b0d54234736e6f9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-12-08 11:24:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4be8f9ba5a306795881fd2a7038a0d20a3dbb50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-08 11:54:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/70/765970/1/OpenVswitch-HW-Offload-os-vif : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"13963fe01ffe30187ab1dd107d81f27a846c6b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 11:56:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45c4dd9f04d14a1ab354e236ea3d90bc : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9636d3f42b54e9bb2f833b582a52a46 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab48772b00504521a7a092d3e480bb58 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a74681795d94c668e6fac0caf1c1d40 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab720b6c97ee44e498c1bdfe908c4655 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b19611739e4e2ea27048d302310477 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/332d547a0f94487183c3fa4667e99127 : SUCCESS in 4m 53s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2a07d33db54248c5bf6dc84eb1d4ac00 : FAILURE in 14m 36s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/60c51b3ceced4c94a313c6c16aefdecf : SUCCESS in 3m 55s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/8628b2cedad84a5caa4623c83c9da447 : FAILURE in 18m 47s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/51f7081fad4f4ecab04dd18aa4e87dc6 : FAILURE in 24m 28s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/c533ef2fe37b4261b0a07d89bff83abb : FAILURE in 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1b67df458cd55e881c87417042fde33ebfdce60d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2020-12-16 08:20:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9f799db2cb9dca56f2a51778b3e229deda0c021c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-16 09:41:43.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/70/765970/2/OpenVswitch-HW-Offload-os-vif : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a5923de900a506c9d0a5639ae64adbf12c2baffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 10:00:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a32854d177c4da5b7971c666fe01475 : FAILURE in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b09c36c2eb49db9a18c7f5d78848fa : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d8bce5df0554d0699ecc3843a7f16ef : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/632a5c7fe95e4095bf78c00becc8f3a5 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eda99783f40040d48d1b26a5dd313ab2 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee156a71d1b4392adcc46ac3672b0dd : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6da584dbca6b493b94acdc9f2a90fbb0 : SUCCESS in 4m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/cdd60e0da61e4552aaa0d4bfbf930c7a : FAILURE in 24m 19s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/29dcbc01fdf143e7be1cf0a690bdc9a3 : SUCCESS in 5m 26s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/8de3ba9877544109b83142ffc2be5c38 : FAILURE in 18m 22s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/5b71ff1778414496bfc32ad3f65b28af : FAILURE in 18m 16s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/0f2d4bd9c19144938989467bd4ee7353 : FAILURE in 12m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a46abc3ea8162bd792b3816aa28305d1ad5e36ac","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2020-12-30 11:54:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b156f7f76238195c3a4a085e3067b1a04b7dc40d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"date":"2021-01-05 07:25:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2070689ef11c0a1ea9dbc0525af8863a0482104a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-05 08:01:45.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/70/765970/3/OpenVswitch-HW-Offload-os-vif : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2ba6c78ae9d37fa7cd0acb1d027c362abce5963b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 09:05:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8ecf8e9b01848f2881b1f325db264d6 : FAILURE in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0c89c53291423bb71d0748b08c2025 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27fd0dc2a1624ad4b956c1b7f2e0620e : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0555c83fc6f4a049f41b549c2ea1bac : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84b5d81c8d2f4d5bb13ee1e7a2745132 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93fe7f306830415a953a6ac0e132a783 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f68e8bc492904bfc95cd24b4d9d312cf : SUCCESS in 3m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b5f0c7a7d8034560b36ad804b14aa17d : FAILURE in 33m 51s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/d0155ca3f5214645907b12f450709c8d : SUCCESS in 3m 40s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/389882feff7b45358447db697da7da91 : SUCCESS in 1h 33m 54s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/4c1ee6ae9f744b58baf75db1f64eacf1 : SUCCESS in 1h 26m 15s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/74082a22b5b84be7ad930e7abd66ea17 : SUCCESS in 1h 30m 43s","accounts_in_message":[],"_revision_number":3},{"id":"0ced74798cafd4de37870f2f1126f4fdab00c27d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-05-24 17:40:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"77e068a7b25ebb6d5bf87a7636073c0a36698337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 19:00:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4945c2e2779543479846ef3390e7d2a6 : FAILURE in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597b05d08dde4835b0fd48c53b9144f3 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a02da3de8594a99a5266b56f8e21632 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b08831eb942f4190a9a852eca82066ae : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5ba266e26634dadace7385b9930950a : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af020da398f41d193efc9f662a097b8 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6292c07b43d1453aa12d656fb6cc8b7b : SUCCESS in 4m 19s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ee90a01a7e5f4f6696af36be9fe4ee77 : SUCCESS in 1h 06m 25s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/ff9be7cfb0eb45f1a646604fcc568f59 : SUCCESS in 5m 12s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/bb9cec59231f41b4b92734911ba6122d : SUCCESS in 1h 04m 13s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/20ea87d1305545b699defe5af5b8750d : SUCCESS in 1h 01m 07s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/29e60c60c733404e97f49d168383abfb : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":3},{"id":"57cfd324cf96afd3995b8cc1b14605bfdf5a1e42","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-24 20:55:43.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/70/765970/3/OpenVswitch-HW-Offload-os-vif : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"3e4395221651598b70e932948b3ca979ad359add","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-03 16:50:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"40c8befb51da1b33eb3367d2dafaf3d421784490","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-03 17:00:37.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/70/765970/3/OpenVswitch-HW-Offload-os-vif : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"34655c759d95e3154f2949ceb7ae2d46b06a87f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 18:24:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2b24044e039423fb52ac4dfe19e0c3b : FAILURE in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cacbd71deb2c4953b57bd34fd2894f70 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bbc3bc3cf794a80922486c54104eaef : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6fa0bedb5ac480fbfffb9dab9c1c31b : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52bfaf45cb904db3aad42a72a9617762 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e232ffaf0cdf432eb697fb914e9179f4 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/619c7ad6e8ec4ed4ab97fd8779e49507 : SUCCESS in 4m 25s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c5ef9ec1f4aa4812adf0f0241d9a4d4d : SUCCESS in 1h 08m 24s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/40d52c3de00841fb96ec2d07111c1e68 : SUCCESS in 5m 02s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/fde995fc0cc0426ca1cc2b87fc96d48c : SUCCESS in 1h 25m 56s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/1313dbeb450e4e3696c854e26166efe2 : SUCCESS in 1h 23m 31s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/421b3c83fce74fdaa519b3951659ea76 : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":3},{"id":"fd9a9a3b04b288908da8651755cd4adb8a3b10f9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-09-07 09:12:07.000000000","message":"Patch Set 3: Review-Priority+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5ac3cdc35227aa9821bcca49322171a0bb4d51dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-07 09:34:02.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/70/765970/3/OpenVswitch-HW-Offload-os-vif : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"1704e87bd92fb757f7fd0b841525d9028dbd466e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 10:44:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b79bcf24363444bb4c2c2c302350b7e : FAILURE in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53801b16c6a340738600f82644232baf : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc794d7b7e484691b13f77539f09c869 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e2cf67f8ab8450295359b94a8698778 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec762a30850145138f79cc24206a14f4 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4b1a1c90e54f638dcaec352cdeeef0 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85f09083f4764315a259167025daa601 : SUCCESS in 5m 34s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/17f9a66b8fd448b88233dda7006be25e : SUCCESS in 1h 28m 54s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/f15d8f60fc134bec872df16fd8dc07c5 : SUCCESS in 4m 25s\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/0cbac48c9a6947858e975c4e8aa63e2d : SUCCESS in 1h 27m 32s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/2f68ac3b1a6e483a8f572dc557c4996f : SUCCESS in 1h 23m 28s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/c152dd725cb448019d3811265daaa823 : SUCCESS in 1h 27m 49s","accounts_in_message":[],"_revision_number":3},{"id":"a6e1a33f6c13cf99ab2e82d840e01a251260c91e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-08 11:25:57.000000000","message":"Patch Set 3:\n\nto be able to merge this you either need to merge\n* the l-c job fix ( https://review.opendev.org/c/openstack/os-vif/+/807839 ) OR\n* drop the l-c job ( https://review.opendev.org/c/openstack/os-vif/+/794211 )","accounts_in_message":[],"_revision_number":3},{"id":"2a40cd6d6c19997e87129d1da739be1702d9a332","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-02-08 16:48:40.000000000","message":"Removed Review-Priority+1 by sean mooney \u003csmooney@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"85a6c862e6b3b603f04496266c241163702b3467","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-01 17:07:13.000000000","message":"Abandoned\n\nstable/train branch of nova projects\u0027 have been tagged as End of Life. All open patches have to be abandoned in order to be able to delete the branch.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3c63ee065a53649c700121a7947c399f23dca95a","revisions":{"39f9bef58e7758164c30a63931f34aaa68c0df12":{"kind":"REWORK","_number":1,"created":"2020-12-08 11:24:53.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/70/765970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/70/765970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/70/765970/1"}}},"commit":{"parents":[{"commit":"7f05023b8389c2184565b4ae9cae82487198bc06","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/7f05023b8389c2184565b4ae9cae82487198bc06"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-12-08 11:24:18.000000000","tz":120},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nTill kernel 5.7 PF and VF representors are exposed as virtual device.\nThey are not linked to its parent PCI device like how uplink\nrepresentor is linked.\n\nStarting from kernel 5.8 due to new change [1] the PF and VF representors are\nlinked to their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\n[1] https://git.kernel.org/pub/scm/linux/kernel/git/netdev/net.git/commit/?id\u003d123f0f53dd64b67e34142485fe866a8a581f12f1\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n(cherry picked from commit b37de19c58c877f5174d76d0a4ba5ab519f464e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/39f9bef58e7758164c30a63931f34aaa68c0df12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/39f9bef58e7758164c30a63931f34aaa68c0df12"}]},"branch":"refs/heads/stable/train"},"00edcb6b639f80bd1b2efe7eb6b6ab35485ce4a7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-16 08:20:50.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/70/765970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/70/765970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/70/765970/2"}}},"commit":{"parents":[{"commit":"c3e25c7e7a871d317b82549f66721d786b51a801","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/c3e25c7e7a871d317b82549f66721d786b51a801"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2020-12-16 08:20:11.000000000","tz":120},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nTill kernel 5.7 PF and VF representors are exposed as virtual device.\nThey are not linked to its parent PCI device like how uplink\nrepresentor is linked.\n\nStarting from kernel 5.8 due to new change [1] the PF and VF representors are\nlinked to their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\n[1] https://git.kernel.org/pub/scm/linux/kernel/git/netdev/net.git/commit/?id\u003d123f0f53dd64b67e34142485fe866a8a581f12f1\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n(cherry picked from commit 8fa7d59be755ffcb61a37b493ed7a8e2b0c8757c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/00edcb6b639f80bd1b2efe7eb6b6ab35485ce4a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/00edcb6b639f80bd1b2efe7eb6b6ab35485ce4a7"}]},"branch":"refs/heads/stable/train"},"3c63ee065a53649c700121a7947c399f23dca95a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2021-01-05 07:25:13.000000000","uploader":{"_account_id":32296,"name":"Mamduh","email":"mamduhala@nvidia.com","username":"Mamduh"},"ref":"refs/changes/70/765970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/70/765970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/70/765970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/70/765970/3"}}},"commit":{"parents":[{"commit":"cda2a020f1ef92ebdd62c592cbad987977f240e6","subject":"Refactor code of linux_net to more cleaner and increase performace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/cda2a020f1ef92ebdd62c592cbad987977f240e6"}]}],"author":{"name":"Mamduh Alassi","email":"mamduhala@mellanox.com","date":"2020-08-12 13:24:47.000000000","tz":180},"committer":{"name":"Mamduh","email":"mamduhala@nvidia.com","date":"2021-01-05 07:24:15.000000000","tz":120},"subject":"Fix - os-vif fails to get the correct UpLink Representor","message":"Fix - os-vif fails to get the correct UpLink Representor\n\nTill kernel 5.7 PF and VF representors are exposed as virtual device.\nThey are not linked to its parent PCI device like how uplink\nrepresentor is linked.\n\nStarting from kernel 5.8 due to new change [1] the PF and VF representors are\nlinked to their parent PCI device, and so \"get_ifname_by_pci_address\" fails\nto get the correct UpLink Representor.\n\nThis patch modifys the behviour of \"get_ifname_by_pci_address\" to\ncheck the physical port name of the netdev in\nvf_pci_addr_path/physfn/net to match the formart for the uplink \"p\\d+\".\n\n[1] https://git.kernel.org/pub/scm/linux/kernel/git/netdev/net.git/commit/?id\u003d123f0f53dd64b67e34142485fe866a8a581f12f1\n\nCloses-Bug: #1892132\nChange-Id: I49f6ae3f0e6bfbf555c8284bfd70371ce90da0c7\n(cherry picked from commit 1226d46cc8bde9531821c68465a05292528142f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/3c63ee065a53649c700121a7947c399f23dca95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/3c63ee065a53649c700121a7947c399f23dca95a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
