)]}'
{"id":"openstack%2Fos-vif~848949","triplet_id":"openstack%2Fos-vif~master~Ia275d1c3a9a9cf63238cf354c8e7d93ed59097a2","project":"openstack/os-vif","branch":"master","topic":"bug/1980948","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-07-07 15:10:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia275d1c3a9a9cf63238cf354c8e7d93ed59097a2","subject":"update ci since linuxbridge is now experimental","status":"MERGED","created":"2022-07-07 11:56:43.000000000","updated":"2022-07-07 15:12:51.000000000","submitted":"2022-07-07 15:10:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848949-bug/1980948","meta_rev_id":"5023497a7a36cedace821be650c4ea2c3d7fec90","_number":848949,"virtual_id_number":848949,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 15:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-07-07 13:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-07 13:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-07 13:37:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-07-07 11:58:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-07 13:05:18.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-07-07 13:16:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-07 13:25:46.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-07 13:37:59.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"df28f72b208a20ce8765f806cfd5e258bc518025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-07 11:56:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6447b2dc41d8925ea1e67f83d7bf2330d4e27574","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-07 11:58:54.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"1c14b2540407b5f8de10cceb1ab462f9bae4099c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-07-07 13:05:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dea0004b6237c95e225825e7f752eb940d2c3f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 13:16:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42978464b15a4f2d9d8f9f0a845a10db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d6c79cd5d78438ba5f68df0c7394353 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df0e42481afe4ba7a255b3d233e1d362 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c00aafdea574096ab6b9ee30d56685c : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7991fff3d19416fa2e4b6d3a23fce9a : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446e10837ad643839e069bc07e19a0dd : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ab9417bbaf44dc1bc3d91d95fbbdf77 : SUCCESS in 3m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1c3b6de8941041ad9a35205aa2cc1699 : SUCCESS in 1h 11m 05s (non-voting)\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/808fd611b7764e7f81ef83fda73f738d : SUCCESS in 3m 40s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/e930e93a58864420befff86c499b9b79 : SUCCESS in 1h 17m 35s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/f9061f68e1694903823ead0191e0f201 : SUCCESS in 1h 08m 56s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/134f0393935b4c1585a55c72ecea8d7d : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c56e94428c453f0559fe44291a8a9e24a1e9d268","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-07-07 13:25:46.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* OpenVswitch-HW-Offload-os-vif http://13.74.249.42/refs/changes/49/848949/1/OpenVswitch-HW-Offload-os-vif : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"67eb6c09d2cd7d070c71a3835c585eff1e9fed1a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-07 13:37:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7721eacbf4c1a41c998fb8a25c8f9a91b06813a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 13:38:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c7925d9ee42fe75bd68e2c2105a6e5efd35bb75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:10:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6e65bf8eaa4f6295717a3185d6fe90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078c631d056740a5a4dcaf341ad405f9 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/106f65f0107b421d92962b0c0208ed36 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/912f3a96b5c04436bbe7f54768f3fa63 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5eee8a613c44d4caabc802a5b4ca8d0 : SUCCESS in 6m 35s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/d804127db83a40ce9fac193d15a2013f : SUCCESS in 3m 30s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/b3ce7a8d176440e5af4996f8ea6ef8dd : SUCCESS in 1h 30m 50s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/b3372de06a59435d9fda25858b78f8f8 : SUCCESS in 1h 24m 28s\n- os-vif-linuxbridge https://zuul.opendev.org/t/openstack/build/c5f8f5d991d44b4896c644a67bd09486 : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":1},{"id":"adbd9e13da5459a8abee159f7b73b87dbadebfee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:10:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5023497a7a36cedace821be650c4ea2c3d7fec90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:12:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0daa2653b6bd40e187cdb2dc3f165ed7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127610c13d1e4961b85e852bd1611773 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"771dfffcd90dcd7c8c95c41744092f5ad4917be3","revisions":{"771dfffcd90dcd7c8c95c41744092f5ad4917be3":{"kind":"REWORK","_number":1,"created":"2022-07-07 11:56:43.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/49/848949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/49/848949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/49/848949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/49/848949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/49/848949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/49/848949/1"}}},"commit":{"parents":[{"commit":"18bd440bbe5692229ac029937000814393898298","subject":"Merge \"Check for hybrid plugging in OVS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/18bd440bbe5692229ac029937000814393898298"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-07-07 11:52:41.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-07-07 11:52:41.000000000","tz":60},"subject":"update ci since linuxbridge is now experimental","message":"update ci since linuxbridge is now experimental\n\nneutron now considers linuxbridge to be experimental\nThis change enables the linuxbridge feature in the\nlinuxbridge job. we do not currently declare the\nlinuxbridge os-vif plug as experimental as it has\nbeen stable for many release and we do not have an\nexperimental concept in os-vif.\n\nChange-Id: Ia275d1c3a9a9cf63238cf354c8e7d93ed59097a2\nCloses-Bug: #1980948\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/771dfffcd90dcd7c8c95c41744092f5ad4917be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/771dfffcd90dcd7c8c95c41744092f5ad4917be3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[]}
