)]}'
{"id":"openstack%2Fos-vif~976359","triplet_id":"openstack%2Fos-vif~master~Iefe8e2c5b77c8144071a0d5600fad34faf2167fe","project":"openstack/os-vif","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-05-02 13:36:30.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-02 09:17:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Iefe8e2c5b77c8144071a0d5600fad34faf2167fe","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2026-02-11 11:36:48.000000000","updated":"2026-05-02 13:37:16.000000000","submitted":"2026-05-02 13:36:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976359-typing","meta_rev_id":"a082d2d7d94a1a24e8a1913229d2911d29f990e2","_number":976359,"virtual_id_number":976359,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-02 13:36:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-05-01 14:35:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-05-01 14:35:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 11:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-25 08:20:04.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"c4ba958189c17a870c947d7684a11864a73482ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 11:36:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89c05db10be5a9bc7e5b1458303cf37068feb1a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 11:49:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f02990d8aa0040d48e08547118a93eaf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ffaa3513ca8444488b3f99f979f352f6 : SUCCESS in 3m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5bc9458b66334c90b1dcca0cc5cc73be : SUCCESS in 1m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bce7ff86182480e88e7e7efb817a41b : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04a40ec987da412ab13caf1a6694dd45 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/798c6ad27d50490e884c554d588bff62 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/846a963a4b894dc9a71d2730fa6c51f0 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/669aa3fc9c56409d86d02890c336da9b : SUCCESS in 5m 18s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/0c872b5cde9945a1aa751956ee3bcc18 : SUCCESS in 3m 41s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/97cdb74bae204efab3383d92b6fd036b : FAILURE in 10m 51s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/f94c76f95eb04da78d19e9ef226876b2 : FAILURE in 9m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cc1667a3a5cf746d2df2b05aac3ed82a145d0f33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 15:16:14.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"70694e9a61b19319960ecac756d1924ef2cbbe5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:04:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/924e1b7258ff465498720624fbb0cc5e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6c76c6f33cee4d80b05d8d463ec0f4bc : SUCCESS in 2m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6bd2a09f9d2d4e0f888125e978726c73 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/519955ecd57542a4b219df80ad80946c : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac1d97b0a0a04bba954baad6c45babf4 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca95b14ee8e14f889efd8ddde41383ed : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd80db7f224b4d46b1eebb5d31f1955a : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e1738097c654cdcabb98327ba27c388 : SUCCESS in 3m 53s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/69b25b25afb843afab9fefa9b8708d7c : SUCCESS in 3m 42s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/353382b175674050a455d84de6d68921 : SUCCESS in 1h 45m 13s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/8848d00d61ed40c498b6af4d283dea84 : FAILURE in 53m 07s","accounts_in_message":[],"_revision_number":2},{"id":"da3f7ed18e53d596aa798450a2606a782f0c383e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-23 16:23:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"42f120e3ac232e702067161dd84f823c70847776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 18:36:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/916b7fca573645f0a58117ca3db1d6e4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7655432ce2014375b8283aebaaa78f71 : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b56b6cf1fa72494e823c6249d0d176ee : SUCCESS in 1m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db555c327964ce88ecf846f912bff46 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/644414c0ad5e4033aa7eb6b9d3ec4737 : SUCCESS in 1m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86b9417000d149cb9576324f19bf94da : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fae3b77bfad4421c9a0565607e54e54e : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb20e30a641a4855b212510d570f1749 : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f602d8f0263f43179e65b23ba7cf426b : SUCCESS in 3m 34s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/eba0efad3f9a4b77938f0e1d6329774d : SUCCESS in 3m 46s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/22a3c8c725524a419561af4e617c9e98 : SUCCESS in 2h 06m 45s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/0d755e0f9a0149278c977cd9cc702067 : SUCCESS in 1h 59m 14s","accounts_in_message":[],"_revision_number":3},{"id":"8ada6cbcc7d5e6776bc6c317fc7f61b13d2a9e3e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-25 08:20:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4422880033e0ffe70424c6cf82de78a8c529052a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-30 15:41:56.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"743ef09938eb8ed7db451cf552a47632612a85df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 15:42:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"faa114dcecf08504e8584de0eee078f4cfc8e223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 16:51:20.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/os-vif/+/976358 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"acb9c32cc5908f594562c3db39e4d407c5df2dbf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-30 17:22:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a604f80fc086e5b21730e8a312ef6a860f5a3d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 19:29:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f80bf5517658435a882c1cfe9c7dcdf4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/407e3f9cae184807a05fcb808e83bc75 : SUCCESS in 2m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/982ebb380ce04e57a4505ad2a6af2bdc : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19c4f34087042138db0edb8a373b211 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14ec470f42554a6d8683009a6b98f41b : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c2b95ad84ef40ae978cb8c91d9c112f : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a970a24d93614f88bdb296c2c346eb59 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6733825b98cb4037bbac23a56241a04d : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dacea1752d04f13878cb041106a00b0 : SUCCESS in 2m 17s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/ddeee21fc004415d8d52f046f2821ce2 : SUCCESS in 2m 46s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/25d15077eb7e4bed88f9be9de62efb8a : SUCCESS in 2h 05m 07s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/d3df21e2bd0b4e82a68bf90b1e3848ea : FAILURE in 1h 15m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b384246183aa49f234a31b289197a07c4a96ba7d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 08:11:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d13b06ca6879c55c8e37e4766601ad9377f37bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 10:19:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed15cf7856bc4c4bb501809831b87bd8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21c148cee10c452db4b7491c95a8cb5a : SUCCESS in 2m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7df783aec08a4beaaa30b8e39e02720e : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940addaf39d74a79b68a65f366ff8600 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6c64349452a48328dc76a075cf4c965 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bd99b6ad06044b8a9f9a967b86a41eb : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/05d6b0db434a4d28a4d67143a913f45a : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1502f37d9eb4453bbf236e4ac65b3d6 : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc702adb93174bf9940a46cc7433dd19 : SUCCESS in 3m 22s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/ecb8227fd9a14ecf82d55724741fdc99 : SUCCESS in 3m 44s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/182b508d8372427dbacce371dbf759ae : SUCCESS in 1h 30m 23s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/eb1a9ba2a239474fb4a81211954d70b0 : SUCCESS in 2h 02m 02s","accounts_in_message":[],"_revision_number":3},{"id":"d7ffcd98bc2b8f94d4b3094981bd110b415bba30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 10:22:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"841e1d7d2994b2eecf0b021d9a5c13612f656bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 12:45:14.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/os-vif/+/976358 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"78e54977fbee3e8ecefa18fd9aab9383e4b46340","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 12:47:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"44309eb476c5a6d3c0d8cc38c047bcad09b46fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 14:17:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/064a7f4f485e43698a8074b260fba90f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd0483db1b3046559b783aca89fb0dc5 : SUCCESS in 1m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5beb3b78cb8b4284b21a87b3f4e0c72c : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91b921cddde64f529a4036431870bdee : SUCCESS in 1m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/247ba8385bdd449da177a53a655a5f06 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e256d5849eb14e678ae98feee1ad61bf : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7678306458264fe7aec44e27f28f64ba : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54073cac5d7f4c6c9db90f3a382e6f18 : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc64594c9b454d1695d4227f83e3db42 : SUCCESS in 4m 31s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/0ccda278736d4f50b5711b8436c93096 : SUCCESS in 3m 59s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/33413edcd9994af6a7a9b03bf4c956af : SUCCESS in 1h 26m 59s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/f59fac1309a34df5bfdc9cb7c16215d2 : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":3},{"id":"2858184879c491e7cb3c43ec15434fcad66536ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 14:27:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"3862784cb865c388cfdaedd69d5a14fd4ff8ff8e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 14:35:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3dd3b445f70de4f2d3971d63687a28aff6adc384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 16:50:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b4023cfed554533a63b4744387fe81a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ce02bf8b3e543a09f7b4b05e6abac0e : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/75757ecda9774c2ba6eaea778f6eb696 : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/220f6432764c45de99c4abe9931672e4 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7798a6c4d86844f78e97873f5cfdecdc : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6883c32426f24781a7809826c1fce46a : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d44b396e4fa4602a8a865d639e24075 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6a2a1a5a6749279ac7d27a3b4cfb11 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3fc819805b74283ba2e6d203ab7a0da : SUCCESS in 4m 12s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/229fc446d3f44f008acb398cac2464ae : SUCCESS in 3m 26s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/0dde156724e543778e6a87ade95a56ee : SUCCESS in 1h 37m 28s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/4fd753639455483ca2041ba2e87b7bcb : TIMED_OUT in 2h 16m 18s","accounts_in_message":[],"_revision_number":4},{"id":"aabd15b7b1ca6fdcce4403ea667df433c8cf3cf3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 20:42:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"46dddd31839a68b8bf01da2c5b86dbc1d1259e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 22:13:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d6da41dce9644b9af01b2b3134168e4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9b55c400efbc453f91c5939f237acc1b : SUCCESS in 2m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bac7d181558d4b939b5dfaea32952afc : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80ccb341f7824476b09890923cf38a0e : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3edef86beb2b4f83bf41c192eb26770c : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fb32e7a51774152a1952c5008aa602b : SUCCESS in 5m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9f470baa8df40b4b228d5ba0a46286e : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac8fe324cf24b248d3f81f772732385 : SUCCESS in 4m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44f9ba0a78ad4d5cbfd3ceab6e24c15d : SUCCESS in 5m 17s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/f2288c09612a482094be68a1e7d2e322 : SUCCESS in 3m 36s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/1b95043903cf4ec794332ea69042e3c2 : SUCCESS in 1h 30m 20s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/52447b80e81e4b148ad13a15ee3e72f0 : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":4},{"id":"d28812082b8bbc12e8e29429e2830402a6c81247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 22:23:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"46119eec7d9828a8cd717169ec024d47127c80b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 00:41:17.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dcfed491b9c47d29ba22309cb5d5555\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e343af888bb1464d85e2a4dd81e2154e : SUCCESS in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/268399bb9e1f4a79b658b83c256085a7 : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c800ea83cd02444baae75e43bd4945c5 : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c2e904626cf45ba9a5734b08707fdcf : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b511feb02db4083a5ca412fbfe5ff96 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8690c79c5de44dc29752304cc9047796 : SUCCESS in 5m 17s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/44b7df81c4b941b0b10663ae093be6ae : SUCCESS in 3m 29s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/c38129b9b0f14742b8c38fb769275951 : SUCCESS in 1h 21m 17s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/c7acfc9120e946948f612621b11040e8 : FAILURE in 52m 49s","accounts_in_message":[],"_revision_number":4},{"id":"eff2ecad9ebf1a408dd0b0f7193811f1eccc5ca5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 09:17:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9af30b3f6345ce467535a971f075ca5bbde8ff85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 11:23:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbf90f0f773547fb8bb796025e417cb8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe0acee3230e4998a13afe3248216300 : SUCCESS in 2m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cb889d86c0f481e8225bc2a445de2ee : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/816922d2227e404694a31422423cd26c : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4605b1c8ec34f6fa1d3423f87ff605a : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfd8b020ffb44873adbd343682b585b7 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e90d71c14614b5d83420b51c69a91cb : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358476d3e02c4ae38a5dd6bbc140ec4e : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d289ddc3e3664542b92dbadc4744401e : SUCCESS in 3m 05s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/3927360185a74390b3d99550f05b0418 : SUCCESS in 3m 34s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/431ffd759910474ca94390319a17c220 : SUCCESS in 1h 59m 16s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/2eb29f294bf248fa8b5c0cbb8075f2fb : SUCCESS in 1h 32m 31s","accounts_in_message":[],"_revision_number":4},{"id":"efc009944a3b9c90ee79187f2377c2ceaa9c4322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 11:24:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9b531e84204437987648824160aa9025fe3a5f54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 13:36:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50304bf18b1f47a1a96bfb89089f91c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3ce266b3abb2462a9e35b91f86766866 : SUCCESS in 2m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa1e41a52f3c4c5aabc21c2a1bf76bf1 : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7157ce17efa24cdcbdae8c77cba733ab : SUCCESS in 1m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ca450e7c7134289a21397b4228d1c78 : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afe22098111148e39a558ec5170981cf : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569d86543a0546f1b3661adf5529005b : SUCCESS in 3m 34s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/9bf6abfe87cd46989b7a684cefe73225 : SUCCESS in 2m 49s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/c5ce40e848974d98a2259a3e4d737d27 : SUCCESS in 1h 58m 01s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/d0f11e616c1a4abb8c3df5295b1ecc28 : SUCCESS in 2h 07m 48s","accounts_in_message":[],"_revision_number":4},{"id":"5d255b8657892fa04d02b5e050707c61ea8013cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 13:36:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a082d2d7d94a1a24e8a1913229d2911d29f990e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 13:37:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9181dedcff51430b9fbdd8094a859ea7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0581a7442b1481faa5641b2f0d57391 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b2ab074e701d1a621e378407f1a5c97e34e55270","revisions":{"f4dc50611350c7d74760e771ea2934776de08b2d":{"kind":"REWORK","_number":1,"created":"2026-02-11 11:36:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/976359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/59/976359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/59/976359/1"}}},"commit":{"parents":[{"commit":"d3b351d1e940a631c90e2966b99d47040f39da25","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/d3b351d1e940a631c90e2966b99d47040f39da25"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:02:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-11 11:07:03.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iefe8e2c5b77c8144071a0d5600fad34faf2167fe\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/f4dc50611350c7d74760e771ea2934776de08b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/f4dc50611350c7d74760e771ea2934776de08b2d"}]},"branch":"refs/heads/master"},"aae1dbd065df80ff7cf0dbcec8b1f6250e51da4d":{"kind":"REWORK","_number":2,"created":"2026-02-11 15:16:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/976359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/59/976359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/59/976359/2"}}},"commit":{"parents":[{"commit":"d3b351d1e940a631c90e2966b99d47040f39da25","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/d3b351d1e940a631c90e2966b99d47040f39da25"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:02:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-11 15:16:13.000000000","tz":0},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iefe8e2c5b77c8144071a0d5600fad34faf2167fe\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/aae1dbd065df80ff7cf0dbcec8b1f6250e51da4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/aae1dbd065df80ff7cf0dbcec8b1f6250e51da4d"}]},"branch":"refs/heads/master"},"35555a09ea6e8fae3daec2a67eacf507e5e573c1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-23 16:23:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/976359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/59/976359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/59/976359/3"}}},"commit":{"parents":[{"commit":"15ba989f0474349ac9bb12b90f07d9dca87b5c8f","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/15ba989f0474349ac9bb12b90f07d9dca87b5c8f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:02:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-04-23 16:17:59.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iefe8e2c5b77c8144071a0d5600fad34faf2167fe\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/35555a09ea6e8fae3daec2a67eacf507e5e573c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/35555a09ea6e8fae3daec2a67eacf507e5e573c1"}]},"branch":"refs/heads/master"},"b2ab074e701d1a621e378407f1a5c97e34e55270":{"kind":"REWORK","_number":4,"created":"2026-05-01 14:27:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/976359/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/59/976359/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/59/976359/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/59/976359/4"}}},"commit":{"parents":[{"commit":"1d1e598de6993e2cb3935a10d3c98131af660bb2","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/1d1e598de6993e2cb3935a10d3c98131af660bb2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:02:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-05-01 14:25:49.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Iefe8e2c5b77c8144071a0d5600fad34faf2167fe\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/b2ab074e701d1a621e378407f1a5c97e34e55270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/b2ab074e701d1a621e378407f1a5c97e34e55270"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
