)]}'
{"id":"openstack%2Fos-vif~976362","triplet_id":"openstack%2Fos-vif~master~I5edd7f05fb71ef19e55c44517d471cebef43b989","project":"openstack/os-vif","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-05-03 11:04:21.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-02 19:05:37.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5edd7f05fb71ef19e55c44517d471cebef43b989","subject":"typing: Add typing to vif_plug_noop module","status":"MERGED","created":"2026-02-11 11:36:48.000000000","updated":"2026-05-03 11:05:14.000000000","submitted":"2026-05-03 11:04:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976362-typing","meta_rev_id":"bf69a50682e81d22144839434ac6da2b1da0fafe","_number":976362,"virtual_id_number":976362,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-03 11:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-05-01 14:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-05-03 08:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 12:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-25 08:30:34.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"84cd394eb12db85bdc4a661cb0d0a85033c01a61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 11:36:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03fc449225c2a22cd8c4d64135ba65b07b47c89e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 12:07:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27c4f0a5ca3e41479e0e324826296270\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d8ab706306e744e7a2da67a6374932fe : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7233990381b1424c9a2345471d055c15 : SUCCESS in 1m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0978bd76d6c4cfbb18f08b9e0e561cf : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b92846f2c0704b92aa33fcded0f9472f : FAILURE in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97855f9cbdf748cab482d49e6c8be1a4 : FAILURE in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be40247cfb64974a3ac882e1cdfd6bc : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94893f1c5d264e38860f412e94b22bb3 : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30ad15a8d1994090a8dd18aee5992565 : FAILURE in 2m 34s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/24a67a9f1ba74c4a9e6b8a3466dd5a39 : FAILURE in 3m 38s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/c20f69e107884517a7184010f004dfad : FAILURE in 21m 08s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/4fa2b545d9ec4e49b4a6281f6100081f : FAILURE in 24m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f24b82b931903d125b62a2f73e13f0d4bd9fcbdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 15:16:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b99067c146b85c6c63ee7cdf2a6952498f5a82d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 15:17:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d2550d31ba98dbf32c6a2192b70d0351a1fa75b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f488249bfd9c43c693886ae0740f80f5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7794c017e97d4d36b696471661f73129 : SUCCESS in 2m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/87d54ba6fdfc4f4eabacddc737d184e9 : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d7bd9ec9d44791ac3d878239c1db3f : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67bd3e7ee0ab4cc8bab1858063ed7892 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e75ddf53d0244e2b8a665b7f88ea5132 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80e4af8e66447ea91d4dda0095d4444 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d29381c66ae04de28fb2277d89d04f8e : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd72ab46b36d417ca9a1433e9874bed6 : SUCCESS in 4m 09s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/e15107f4094b4d4d9714760633608e2a : SUCCESS in 2m 49s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/5c23785e241442b4a28e283c723c372d : SUCCESS in 1h 48m 07s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/464a84822f264963a877a4ec88eb0fb3 : SUCCESS in 1h 54m 52s","accounts_in_message":[],"_revision_number":3},{"id":"fc80a4fe4e649adbf7b0cbaea24599a9791cddb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-23 16:23:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"88d5b60318539589fa15609fee31bd4fd812c4a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 18:46:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68f22aa014ff46828598b5b1211e109e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff652f84b231472d9c2bec83019af6ab : SUCCESS in 2m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a0a6d461c4d4da6b9d6042e7b0f7806 : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67685c39bf6040c8b314c65be551120a : FAILURE in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df2ab4c6d68d491c9b1c515156788345 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5f3feee661d4b4e83103a28917feec4 : SUCCESS in 1m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69b58bb8f7b1452c85d9f17d46394a3b : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a21009bde0441708e07053d1c80d4d4 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea78c16c37794628a8b8e6bb6a51ac55 : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4751d452ea74f77a9333190b0ec18b6 : SUCCESS in 3m 49s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/7af1edc5214042659e41725191d3423d : SUCCESS in 4m 17s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/3fe21db7d72c424482bbf76ed3ede5c8 : TIMED_OUT in 2h 15m 49s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/6e8bab2bdee64a33a118ca7a1b3e91b4 : SUCCESS in 1h 47m 57s","accounts_in_message":[],"_revision_number":4},{"id":"1b17532e91bf3b65698d8e8d305caf879c7c9d92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-24 11:06:43.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"c28c3e6ef077208c593799df8eb1b1bc01eba852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 13:14:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b3c410b9e0b48afba35479e8b51b38b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd272c48ddb044fdb0b254191edec014 : SUCCESS in 4m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4de77018265e4bd5a8a4363c335bcb2b : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8cb8d2738664a218a5507d7da04d5fc : FAILURE in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd6d41c8dbaa4b929151737f78d447eb : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a25e4927d86469e84f1d22a90c2b099 : SUCCESS in 5m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bf8bdf2b6ea8434ab0ff5e5327269596 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc53830e8a0242a5b2107d8d126a9153 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8e9e88c97d84b908013d5706d7ecda9 : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeec138ffa134a1b9e5ea926691510f1 : SUCCESS in 5m 23s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/b36006f5b14f436a8c24b4a8e47816eb : SUCCESS in 5m 12s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/e469755889e5413a99b80fe6650958c8 : SUCCESS in 2h 00m 10s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/32c15035177e4e8aaca968bb20576fdd : SUCCESS in 2h 00m 34s\n\nWarning:\n  Comments left for invalid file os_vif/__init__.py","accounts_in_message":[],"_revision_number":5},{"id":"03472451ac7656c3a3c998a2b909f9ce022a68f2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-04-25 08:30:34.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a45b4c4fd80c60683bdfe339c8aef2a54bb96822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-27 11:20:19.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"1a54a904e738f247a5e49ca24e473a44c76e06c0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-27 11:21:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4045192a280f5be4ef2d91228ed40773ce84e12a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 13:21:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/698cabe1bd7545dd8bb07af27c518974\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9050365544cf4cda862aefe66ffaa189 : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3f738528f344c9da6477f8051175ed3 : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29a9a678f514cc099196c3c41fea27a : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b08f86efe654deea1e4dbd1d8014518 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af56f2b322f041b28c111e6556e11cfe : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5e0227a372904cd0a4d2eeacba7c7720 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e07c06467824df18eaafa42a82c93d7 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/988b2c2e76034e94b9e001fb18a71443 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93104deb3bac4a6a9961d8615b57a6d4 : SUCCESS in 3m 44s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/4e0420787aa04f8795f148e73d68d840 : SUCCESS in 4m 01s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/d880426881404d21b739647435669c97 : SUCCESS in 1h 55m 15s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/d58203972dc44e69bdf4e5f3125c61e2 : SUCCESS in 1h 45m 10s","accounts_in_message":[],"_revision_number":6},{"id":"d5f960201534a87e80dad5cb4adf63cfa114fffd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-27 13:27:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b869cb56651c411b36ebc5d3c04029145372ea20","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 08:12:00.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d05dd4bafb3029f7b2e9286c05c1f2f23beb2619","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 10:22:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"97ae56aa391bdf88fc10b61f0a906c9379ed4a19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 12:45:17.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/os-vif/+/745162 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"c1046500fed4bac2a471cf66dbe1a0968fe8fb86","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 12:47:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"457fd2809c7e260cb7f11abda257228f9baf954a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 14:27:22.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"f5331af1840184e346ca92fb2901e01d6a7734dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 16:39:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84fe8d7c1fdb496d9fa8b604416d2a31\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/43072342d5ec48de9aa39fa45ad655c9 : SUCCESS in 2m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0f99c5b276a4b418a12314a5b8d9ebe : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e25e6fa572948fa977a4f4dbd696221 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdd57d6447c5452d8f6d0ba4c67d249d : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10242fc104544198bbb657b808aaee2e : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/611e46b6e6cd422cb1bd8513be8c1c12 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f94923bcde4c23a796a67d0bcbf1fe : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f59d94c8c1454aea999389ef51b5a7b1 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ada2b44320ec420d928987a691bc6243 : SUCCESS in 3m 30s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/87de22f343d24a69a266530e39fa8399 : SUCCESS in 3m 44s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/a95d8444403c48e19b1abc5107fa01d4 : SUCCESS in 2h 06m 33s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/530f64de4cee4dbe86193a0ba79c2c84 : FAILURE in 1h 28m 57s","accounts_in_message":[],"_revision_number":7},{"id":"46d6220f95a62219487f3b667a995af45595dff5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 12:29:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"76810cd17606c9f13c98152f00e101687f2543aa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 12:35:09.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"2901466a3a7b659d795cb03f8ff13189ff802005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 14:34:18.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/065b16f435d44e849336872bd0caf211\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6c59fc6704334778ac9a5cd64f503d4b : SUCCESS in 1m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c3ee740b2e124c3197208e451809634c : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4461d1c18d64f8e956f61aadbf77303 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c806cae16c94c31abc6394e5f15fb5a : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d5e2fa4de37482fa2712d5db5d1c751 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f536708a1f54fc69673826c06b73563 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d14bdea3ef3645819496696453852414 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fba6c0074bbd4d099a38e139615e3d55 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d243f7a2204304855d874ca104619b : SUCCESS in 3m 49s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/d8ede1d7c68e495cbf3c1bb98e187d9e : SUCCESS in 2m 32s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/36ddac6ef6624225a89c01864b1dc435 : SUCCESS in 1h 59m 13s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/79ec9b1a6eb348688da904c04b9287ec : FAILURE in 1h 16m 39s","accounts_in_message":[],"_revision_number":7},{"id":"298e0a3ae38fc24598f8ab4d614e53f01a6d811d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 14:35:15.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"53aaf7fe71735f5b826700c335fc0ab8d3188231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 16:13:05.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ea0d6a4fa524cf5bd1e14404fcee9aa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/61e89acdd0964402b5027914720189ac : SUCCESS in 2m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d7593752c6404d968444650253d86573 : SUCCESS in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344629820ee8453cbe748d90053d0b83 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d6fdd22b67741138278f24c24f937db : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf06293c829b4c15a6c1cbfdad9d15d7 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e1f70befcff94d90978ed002d3843386 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c0408414604b118cb3c2bfb8b2a77c : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/901864058e024ea38fa81966a6f77af2 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1dc31d404d34fa8802eb7510f2d848c : SUCCESS in 2m 10s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/4bd60ce5ca7843d0b71ab7b710c4e38d : FAILURE in 4m 42s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/c6b1f7723c8040b5a7bc6f4cbfac5489 : SUCCESS in 1h 27m 59s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/db8b4b34a4564b87a6c195145d4c7ae4 : FAILURE in 1h 11m 07s","accounts_in_message":[],"_revision_number":7},{"id":"0dc4241fe01d74de37e2dc76341d33d2bc358a6b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 16:35:22.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"56d96043cb9fb1c3d164f219735a3dd1e717c820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 18:59:58.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85426baa5c354228b05c98eef88d54c8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/45b2ed9de00442ff85b8dcb28eb45bee : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac1c936acc9f4a1490468d4f124583c0 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c45c5e60209c43ecaf4e680c0b0c4740 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/193da05592184f088a24a07784653dd0 : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c0d7a5107c340f0935859dc923b460f : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/800a475744c64567991ba5c656685a63 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2000643fcf864080ac1a8c28644e6ef3 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67fd11cd10da407d9b29877661ce2626 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9edc426139f540bea620a123f4d0c563 : SUCCESS in 3m 37s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/d94b6d40a13a42e89a1c5596648d3428 : SUCCESS in 3m 24s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/c8052cb4da83451db327f9f8d9ed19c3 : TIMED_OUT in 2h 16m 42s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/d05efcf0ed5f41b780b694526e71409d : SUCCESS in 2h 06m 17s","accounts_in_message":[],"_revision_number":7},{"id":"36c6d0ddc14318133c8934c9210ae04cd8cd088f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 19:05:37.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"d44961e2e29281459e1a22e9259c1437b51ea409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 21:12:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f32ee6c8afaf4311b55ff1ae865c3339\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/df00338a5b5944d3bfcf7554d8c03786 : SUCCESS in 2m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/299376a195be4850a1be51cb045349fe : SUCCESS in 1m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876a06089ea24350858cfa4eba5edb92 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a0c5acb621346b5975c1ecc4ba2ac9a : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31ccf859e2fd4c25b99408e9957724e2 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6672106d5eb448e9891d7f585d6ca5f0 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d0c502628146538133300db85d4f9b : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a8baa506f0544318462e0b5218ed5d0 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04f01c9ad6de4ee1a5c64ebacfad9a90 : SUCCESS in 3m 37s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/37956bf244294286a080a4fe139fd121 : SUCCESS in 2m 56s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/fed9690df1d943ab9b3d9b549584c376 : SUCCESS in 1h 56m 37s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/050dc7b7b69945d4abc358715f54496f : SUCCESS in 1h 59m 15s","accounts_in_message":[],"_revision_number":7},{"id":"6af2647a7554e447baaf84b011024d88729d086b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-03 08:59:58.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2506feee9d24180a0dc6427c93ab0dd9d80aa1a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 09:00:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b33372ee171f1f374deef7aba84db09c62f74817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 11:04:21.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d12bd1c09e4818a7f0296376655f20\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ea78f09abddd414dbf790034f09e4b45 : SUCCESS in 2m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/63e7f48e11ed4449b55e681bc32fbda7 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a368770bd9894922b66c57a4cc4db2f8 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfccb62dcc154c03aca1bfa0334e905b : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f958f3e0d984131b5d180abef00a73a : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baae6767611f409fb1bb409a697990ea : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1906e33c55e43c1a84961cc6d405ee1 : SUCCESS in 2m 17s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/fed9cb95c2fd4f308938abab31b366e5 : SUCCESS in 3m 31s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/959d4f711d0742c0b4f3a0822ba78a06 : SUCCESS in 1h 15m 46s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/bd01fcc8cb4843e485bd667c88d9cf7c : SUCCESS in 1h 59m 19s","accounts_in_message":[],"_revision_number":7},{"id":"2c25270bacaeac1b58cac92c9261f438951a8ebe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 11:04:21.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":7},{"id":"bf69a50682e81d22144839434ac6da2b1da0fafe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 11:05:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/031844abd88149f0ae3ae65457abc65e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c768b43b38804a3a8413b49d7fc68ce8 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4607d559f65f471aaf282d6f78b70edc : SUCCESS in 40s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"22dcd8d18a08bc9cacfe3deeb79ee1072e17bc9d","revisions":{"71d75b62b4ec3cc844e1c3bbb912377d61f103a2":{"kind":"REWORK","_number":1,"created":"2026-02-11 11:36:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/976362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/62/976362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/62/976362/1"}}},"commit":{"parents":[{"commit":"3a0462dde21961ad7ce2f992d1e0a5372ee465c7","subject":"typing: Add initial type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/3a0462dde21961ad7ce2f992d1e0a5372ee465c7"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:33:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-11 11:19:34.000000000","tz":0},"subject":"typing: Add typing to vif_plug_noop module","message":"typing: Add typing to vif_plug_noop module\n\nChange-Id: I5edd7f05fb71ef19e55c44517d471cebef43b989\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/71d75b62b4ec3cc844e1c3bbb912377d61f103a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/71d75b62b4ec3cc844e1c3bbb912377d61f103a2"}]},"branch":"refs/heads/master"},"56a75f1355f8f18f1f52a8128deb3d93fed8b9fa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-11 15:16:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/976362/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/62/976362/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/62/976362/2"}}},"commit":{"parents":[{"commit":"6fc63d81ab93cf0ec240bf3cbbbde373c7628c69","subject":"typing: Add initial type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/6fc63d81ab93cf0ec240bf3cbbbde373c7628c69"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:33:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-11 15:16:35.000000000","tz":0},"subject":"typing: Add typing to vif_plug_noop module","message":"typing: Add typing to vif_plug_noop module\n\nChange-Id: I5edd7f05fb71ef19e55c44517d471cebef43b989\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/56a75f1355f8f18f1f52a8128deb3d93fed8b9fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/56a75f1355f8f18f1f52a8128deb3d93fed8b9fa"}]},"branch":"refs/heads/master","description":"Rebase"},"3d75ce11ae0f7902a1f0f0641ef76ff121526304":{"kind":"REWORK","_number":3,"created":"2026-02-11 15:17:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/976362/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/62/976362/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/62/976362/3"}}},"commit":{"parents":[{"commit":"6fc63d81ab93cf0ec240bf3cbbbde373c7628c69","subject":"typing: Add initial type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/6fc63d81ab93cf0ec240bf3cbbbde373c7628c69"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:33:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-02-11 15:17:31.000000000","tz":0},"subject":"typing: Add typing to vif_plug_noop module","message":"typing: Add typing to vif_plug_noop module\n\nChange-Id: I5edd7f05fb71ef19e55c44517d471cebef43b989\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/3d75ce11ae0f7902a1f0f0641ef76ff121526304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/3d75ce11ae0f7902a1f0f0641ef76ff121526304"}]},"branch":"refs/heads/master"},"417e46084bf15bf9aa8e4d463e8a54c7de858c25":{"kind":"REWORK","_number":4,"created":"2026-04-23 16:23:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/976362/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/62/976362/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/62/976362/4"}}},"commit":{"parents":[{"commit":"37105a330e275f8ad892ae1162d157bdbdcec44a","subject":"typing: Add initial type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/37105a330e275f8ad892ae1162d157bdbdcec44a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:33:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-04-23 16:23:24.000000000","tz":60},"subject":"typing: Add typing to vif_plug_noop module","message":"typing: Add typing to vif_plug_noop module\n\nYes, this is dead simple: that\u0027s the point :)\n\nChange-Id: I5edd7f05fb71ef19e55c44517d471cebef43b989\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/417e46084bf15bf9aa8e4d463e8a54c7de858c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/417e46084bf15bf9aa8e4d463e8a54c7de858c25"}]},"branch":"refs/heads/master"},"6a360a61eeaa51ebbffefef480e2c326596b9a4b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-04-24 11:06:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/976362/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/62/976362/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/62/976362/5"}}},"commit":{"parents":[{"commit":"96d29cdd910e5e1b62b10a98e922e5ce8f68a84b","subject":"typing: Add initial type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/96d29cdd910e5e1b62b10a98e922e5ce8f68a84b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:33:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-04-24 10:04:15.000000000","tz":60},"subject":"typing: Add typing to vif_plug_noop module","message":"typing: Add typing to vif_plug_noop module\n\nYes, this is dead simple: that\u0027s the point :)\n\nChange-Id: I5edd7f05fb71ef19e55c44517d471cebef43b989\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/6a360a61eeaa51ebbffefef480e2c326596b9a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/6a360a61eeaa51ebbffefef480e2c326596b9a4b"}]},"branch":"refs/heads/master"},"4baff1aaf5b2dd4fe820199e54ecff92811082df":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-27 11:20:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/976362/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/62/976362/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/62/976362/6"}}},"commit":{"parents":[{"commit":"231c814d643f9fce37e7e8c3ecb1aa1122545752","subject":"typing: Add initial type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/231c814d643f9fce37e7e8c3ecb1aa1122545752"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:33:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-04-27 11:20:11.000000000","tz":60},"subject":"typing: Add typing to vif_plug_noop module","message":"typing: Add typing to vif_plug_noop module\n\nYes, this is dead simple: that\u0027s the point :)\n\nChange-Id: I5edd7f05fb71ef19e55c44517d471cebef43b989\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/4baff1aaf5b2dd4fe820199e54ecff92811082df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/4baff1aaf5b2dd4fe820199e54ecff92811082df"}]},"branch":"refs/heads/master"},"22dcd8d18a08bc9cacfe3deeb79ee1072e17bc9d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-01 14:27:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/976362/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/62/976362/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/62/976362/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/62/976362/7"}}},"commit":{"parents":[{"commit":"960b217b57e3b6f29bd28c033a2074477e0ce0a8","subject":"typing: Add initial type annotations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/960b217b57e3b6f29bd28c033a2074477e0ce0a8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-29 14:33:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-05-01 14:25:51.000000000","tz":60},"subject":"typing: Add typing to vif_plug_noop module","message":"typing: Add typing to vif_plug_noop module\n\nYes, this is dead simple: that\u0027s the point :)\n\nChange-Id: I5edd7f05fb71ef19e55c44517d471cebef43b989\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/22dcd8d18a08bc9cacfe3deeb79ee1072e17bc9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/22dcd8d18a08bc9cacfe3deeb79ee1072e17bc9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
