)]}'
{"id":"openstack%2Fos-vif~989113","triplet_id":"openstack%2Fos-vif~master~I978cd2d6823c0346deab2ba386261cb97589fcea","project":"openstack/os-vif","branch":"master","topic":"gate-fix","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-19 15:11:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I978cd2d6823c0346deab2ba386261cb97589fcea","subject":"zuul: Disable cinder jobs","status":"MERGED","created":"2026-05-19 12:07:33.000000000","updated":"2026-05-19 15:14:18.000000000","submitted":"2026-05-19 15:11:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989113-gate-fix","meta_rev_id":"7b3ae78439363e481632d50d214ffd7b61b576fc","_number":989113,"virtual_id_number":989113,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 15:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-05-19 12:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-05-19 12:12:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-19 12:12:46.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-05-19 13:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fa5b52b5cee41a45d50654688fe648e5f767933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-19 12:07:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cf2f975eb766533fe370e4a8915e8ee6fb3e851","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-19 12:12:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2bc53d940befedcfe0ba3084850ce2a5f3c11e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 13:40:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/172346e51bcc46dfbcf7d234e21fba2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dcc773ce51e4291a747b58b71a0abd3 : SUCCESS in 1m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bcede14e6c47484c8f366faec4286a6b : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0545d080680413985281c6991274617 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c9d092852e04f23a8a0ac04cbf527fd : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8174d296ed147feb21adb20181bf37a : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e56dce0bd7924db8b36baf1fb1509ae5 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97128d0c784b41e38807aff3ea876872 : SUCCESS in 3m 35s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/b0d7067061cd482a86195abbc3378ec8 : SUCCESS in 3m 43s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/fbd7f9784a7148e780da7450aa06ea89 : SUCCESS in 1h 26m 56s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/56eb3d3ba5974f9081d556098184e42b : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b8dff93e865c229cb759d420287c1f892ef1bd24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 13:41:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7aeaef69db2cd02685e3e11fc6d14f7fcb83dd3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:11:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e34479a3fc741edabfc8f767bf3843c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dbbe098e60345ce8640fb3a91218789 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4563352bf8c14068a3708c3ad5b6f9e1 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e08ec8363d82410ba9aae05685710a9d : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/620043e1fbbc4fdd91f58ca2112c9bb9 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44f9f0a07e2a4067b26ca640885182d9 : SUCCESS in 2m 13s\n- openstack-tox-functional-ovs-with-sudo https://zuul.opendev.org/t/openstack/build/863ccc113ae04e5ca99eab9dceaf4801 : SUCCESS in 4m 08s\n- os-vif-ovn https://zuul.opendev.org/t/openstack/build/aff5bbad65474c3ba3f4b1ee5942f274 : SUCCESS in 54m 47s\n- os-vif-ovs-iptables https://zuul.opendev.org/t/openstack/build/c24b9dcf42fb45f9802ce501f79ac515 : SUCCESS in 1h 27m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a08ae7116821288e882aa2ca97feedb2b20b72c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:11:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7b3ae78439363e481632d50d214ffd7b61b576fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:14:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec50a52ccc714b819fb52a3211b06781\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fe22032ea0433e9dec05e06223d3c8 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4676576f0ae5c94937c07b9de01baa560e9fa14","revisions":{"f4676576f0ae5c94937c07b9de01baa560e9fa14":{"kind":"REWORK","_number":1,"created":"2026-05-19 12:07:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/989113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-vif","ref":"refs/changes/13/989113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/13/989113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/13/989113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-vif refs/changes/13/989113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-vif refs/changes/13/989113/1"}}},"commit":{"parents":[{"commit":"89a56221c5995cc51aa3d4646eff5f5a0b98313e","subject":"Merge \"Drop support for Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/89a56221c5995cc51aa3d4646eff5f5a0b98313e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-05-19 12:07:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-05-19 12:07:02.000000000","tz":60},"subject":"zuul: Disable cinder jobs","message":"zuul: Disable cinder jobs\n\nWe are testing neutron and nova interactions. We don\u0027t need Cinder in\nthe mix.\n\nChange-Id: I978cd2d6823c0346deab2ba386261cb97589fcea\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/f4676576f0ae5c94937c07b9de01baa560e9fa14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-vif/commit/f4676576f0ae5c94937c07b9de01baa560e9fa14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
