)]}'
{"id":"openstack%2Fosc-lib~1003372","triplet_id":"openstack%2Fosc-lib~master~I10f3c557a79251023cf553679b3e3620ce9c622c","project":"openstack/osc-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-01 21:02:39.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-01 22:20:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I10f3c557a79251023cf553679b3e3620ce9c622c","subject":"Fix typo","status":"MERGED","created":"2026-09-01 19:59:30.000000000","updated":"2026-09-01 22:21:04.000000000","submitted":"2026-09-01 22:20:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003372","meta_rev_id":"51c9512bf01b9b27925cc98d940aa5e5dd913e26","_number":1003372,"virtual_id_number":1003372,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 22:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-01 21:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-01 21:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 20:00:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-01 21:35:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2a8936495e8cb67900eff74cd6734ba760623bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:59:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c14c72cb3fa04eae7db0768afb0966b377274ea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 20:00:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"91818b9239ebee372aa347bc3ef37b794f4ce163","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 20:00:17.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"e6a9ada415ee60917b3e3502927ac1647649e215","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-01 21:02:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7177852df78aeec5c8021012824ec6c9c999d45f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:35:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfe737cca6304d478d84b7a59f7d5818\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498782e480994835852c1422195671ac : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29e8d5e85ef84626ab97bd832c77b6d5 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b53d807cabf4371aa1a444d7600791d : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6633b80e9da74717a9dbe99016413cba : SUCCESS in 4m 39s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/25d346f574fb4767bb52cc7d573de36b : SUCCESS in 8m 48s (non-voting)\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b428462a23114ea6a087573059dd2d0b : SUCCESS in 4m 40s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/cba774f62f5b47d185379e69da9c05b2 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6c5d7738c04b158bb0c110bef7fc64 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/763877f6117d4a3f888f93110ab36a73 : SUCCESS in 3m 14s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/5f7ceb441cc9428d86b69a7a0befb00c : SUCCESS in 57m 51s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b05c5c65dccc416184ca2a8b04eb7bbb : SUCCESS in 1h 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c990d110f6c94c12df1c0a957fcfb0918daddec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:35:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b6614825c4ab3bf792ac95ca4ec176f7fad6e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 22:20:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d448beaf2d9f4e499b93f775b2f2f91c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f62ddaff4b41d3a20754e784da83de : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e12b16dc05df48368a46cefbcf8c28a7 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4acdefdef35a4f92a4555bb8d49e8124 : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/caa773bc66074b8ba15db5b204f820b0 : SUCCESS in 3m 26s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b0552a79df294aefaa2c2b43ec2a7baf : SUCCESS in 2m 40s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1c2415e3be77406c99cbcb8dd26128ae : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ae22757aec47f79a7e8067964c6400 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/927905c2f1924743965d237aafd8d7f3 : SUCCESS in 6m 02s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/726c1f7800f24db18f5e3e52ba938c10 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":2},{"id":"162f8cec59134db96e95ea6b85025c62bbdbb3c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 22:20:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"51c9512bf01b9b27925cc98d940aa5e5dd913e26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 22:21:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeb3a7761ecb4693ad77080906a6aab8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c8b50dd2d04e6c8e925f5dd0ea3928 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2aa19a0730346ba28572f4d1add7bd2cfb1fc2e","revisions":{"5172afbb60df0c68b9b1505c139018929a35a5c0":{"kind":"REWORK","_number":1,"created":"2026-09-01 19:59:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/72/1003372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/72/1003372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/72/1003372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/72/1003372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/72/1003372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/72/1003372/1"}}},"commit":{"parents":[{"commit":"e588cefbe243dca28b78ec782516554ba8aa951e","subject":"Merge \"Add assertIsSubset helper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/e588cefbe243dca28b78ec782516554ba8aa951e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:59:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:59:30.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I10f3c557a79251023cf553679b3e3620ce9c622c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5172afbb60df0c68b9b1505c139018929a35a5c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5172afbb60df0c68b9b1505c139018929a35a5c0"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"b2aa19a0730346ba28572f4d1add7bd2cfb1fc2e":{"kind":"REWORK","_number":2,"created":"2026-09-01 20:00:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/72/1003372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/72/1003372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/72/1003372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/72/1003372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/72/1003372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/72/1003372/2"}}},"commit":{"parents":[{"commit":"e588cefbe243dca28b78ec782516554ba8aa951e","subject":"Merge \"Add assertIsSubset helper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/e588cefbe243dca28b78ec782516554ba8aa951e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:59:30.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 20:00:11.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I10f3c557a79251023cf553679b3e3620ce9c622c\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/b2aa19a0730346ba28572f4d1add7bd2cfb1fc2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/b2aa19a0730346ba28572f4d1add7bd2cfb1fc2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
