)]}'
{"id":"openstack%2Fosc-lib~512699","triplet_id":"openstack%2Fosc-lib~master~I6cf4e22d676f4a55f84863125aee7318915a7404","project":"openstack/osc-lib","branch":"master","topic":"bug/1724283","hashtags":[],"change_id":"I6cf4e22d676f4a55f84863125aee7318915a7404","subject":"Allow to use the none auth plugin","status":"MERGED","created":"2017-10-17 16:26:55.000000000","updated":"2018-03-15 23:32:51.000000000","submitted":"2018-03-15 23:32:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"512699-1521156771244-58992c28","meta_rev_id":"6aa3a8b9f38e7ae0d64a77377e5d13950ee6f0be","_number":512699,"virtual_id_number":512699,"owner":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-03-15 09:45:11.000000000","_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2018-03-15 23:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"recommended":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":1,"date":"2018-03-14 07:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2018-03-14 07:28:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":2,"date":"2018-03-14 16:15:28.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":1,"date":"2018-03-15 13:46:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"all":[{"value":0,"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":1,"date":"2018-03-14 20:44:33.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-16 18:14:04.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2018-02-20 16:53:33.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"},{"updated":"2018-03-14 07:29:16.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2018-03-14 20:44:33.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2018-03-15 13:46:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2018-03-15 23:32:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8d91de77bc1d7699650348e9d1a54040d761374","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-17 16:26:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"068ed4e9e26691249f5748c8edce03ff664fad31","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-17 16:29:28.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b4453d3fd82c272449763dda28ec5cc22194afce","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-17 18:20:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14ea8bea86f3d03eaf9ff0952c794b662e8189c4","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-17 18:22:55.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a762801601f316c901f234b1ab32e9f9aa69f547","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-17 18:32:32.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c3e21358e1679d5e25447daa2b4abf1082ebfb9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-17 21:21:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/3/check/openstack-tox-pep8/75ba7a3/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/3/check/openstack-tox-py27/6ee20f0/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/3/check/openstack-tox-py35/12241c7/ : SUCCESS in 5m 20s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/3/check/openstack-tox-pypy/e1c69e0/ : SUCCESS in 4m 30s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/3/check/legacy-tempest-dsvm-neutron-src/b919665/ : SUCCESS in 1h 22m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/3/check/build-openstack-sphinx-docs/1e9a45a/html/ : SUCCESS in 4m 33s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/3/check/build-openstack-releasenotes/3b150f5/html/ : SUCCESS in 4m 13s\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/99/512699/3/check/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/d82d29f/ : SUCCESS in 2h 03m 40s (non-voting)\n- legacy-osc-dsvm-functional-tips http://logs.openstack.org/99/512699/3/check/legacy-osc-dsvm-functional-tips/852581b/ : SUCCESS in 56m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8da6fabf683e2b43ee4e9e1afaecd0485b070c1b","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2017-10-25 20:19:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThe proper way to do this is to use an actual KSA plugin that does the right thing rather than hack around trying to avoid plugins.  The --os-endpoint option is declared in KSA\u0027s admin_token plugin which is actually rarely used so you can not depend on that option being available from that source.\n\nI think the proper approach would be to use KSA\u0027s NoAuth plugin, or a subclass of it that defines the option that you want to get the service endpoint from (it could be --os-endpoint or --os-url if that doesn\u0027t confuse users.\n\nThis really should be done in KSA itself, Jamie still gives me grief about defining the TokenEndpoint plugin in OSC.  Putting it in KSA will make it available to all libs/apps that consume it.  If it can\u0027t go there for some reason, putting it here should look a lot like the openstackclient.api.auth_plugin that has the TokenEndpoint class, but in osc-lib.","accounts_in_message":[],"_revision_number":3},{"id":"ad02235dad5036c70b23cf2db0ff9ca6725c5a2d","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-27 14:44:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0b259c4ca827cf0fe510a19fcf9cc88e4f07c169","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-27 14:44:23.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"62f3eaf40b21b89c3ae8137c03a016090e8193fd","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-27 15:00:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0319f1c34fc1c0eddae641f87fea94b3b1337968","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-27 15:00:03.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"a82bd9beaf7c932b3a3ee3b89ff128011f6442e6","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-27 15:02:28.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"f46bac3faa1b4b1840c9457885001d1c0fd1a076","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-27 15:04:13.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"b3191e11faa2d3a1f67cb59d1759ea57e73ebed5","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2017-10-27 15:46:46.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1d82d8f3a56b4c5d397c0936713db24b53e1576e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-27 17:43:52.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/6/check/openstack-tox-pep8/20258b3/ : SUCCESS in 6m 41s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/6/check/openstack-tox-py27/01196a2/ : FAILURE in 6m 33s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/6/check/openstack-tox-py35/3f6e198/ : FAILURE in 4m 59s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/6/check/openstack-tox-pypy/5ce73aa/ : FAILURE in 6m 49s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/6/check/legacy-tempest-dsvm-neutron-src/174b510/ : SUCCESS in 1h 45m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/6/check/build-openstack-sphinx-docs/71f84dd/html/ : SUCCESS in 6m 26s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/6/check/build-openstack-releasenotes/f090f52/html/ : SUCCESS in 5m 34s\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/99/512699/6/check/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/61d0964/ : FAILURE in 34m 06s (non-voting)\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/6/check/osc-functional-devstack-tips/7384048/ : SUCCESS in 47m 19s","accounts_in_message":[],"_revision_number":6},{"id":"e94ac172831e3815fca25f185dbcc87f6f80aa65","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-01-29 17:58:49.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"2adafbe6bfc42167b70af79b7197ff9a4b94b1dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-29 20:58:23.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/6/check/openstack-tox-pep8/3f16f18/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/6/check/openstack-tox-py27/1eeea81/ : FAILURE in 4m 22s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/6/check/openstack-tox-py35/001af09/ : FAILURE in 4m 03s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/6/check/openstack-tox-pypy/9514a28/ : FAILURE in 4m 23s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/6/check/legacy-tempest-dsvm-neutron-src/f14ace1/ : SUCCESS in 1h 34m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/6/check/build-openstack-sphinx-docs/9f31f11/html/ : SUCCESS in 7m 12s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/6/check/build-openstack-releasenotes/1bb0366/html/ : SUCCESS in 4m 07s\n- osc-functional-devstack http://logs.openstack.org/99/512699/6/check/osc-functional-devstack/d8ec47a/ : SUCCESS in 41m 29s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/6/check/osc-functional-devstack-tips/a652d43/ : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"66f24509b921b3232a7ea44d8a6aedd0d2569b46","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-02-01 02:02:22.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"bf5ae03a69c14275b9e897ed7402c7e88db7e656","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 04:25:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/7/check/openstack-tox-pep8/b39e03e/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/7/check/openstack-tox-py27/3afbc15/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/7/check/openstack-tox-py35/90a5287/ : SUCCESS in 3m 53s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/7/check/openstack-tox-pypy/20416be/ : SUCCESS in 6m 17s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/7/check/legacy-tempest-dsvm-neutron-src/21bae49/ : SUCCESS in 1h 34m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/7/check/build-openstack-sphinx-docs/d994573/html/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/7/check/build-openstack-releasenotes/5765544/html/ : SUCCESS in 3m 18s\n- osc-functional-devstack http://logs.openstack.org/99/512699/7/check/osc-functional-devstack/101436f/ : SUCCESS in 41m 57s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/7/check/osc-functional-devstack-tips/23d9d97/ : SUCCESS in 45m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"72b39dcac87ceb5361ad48e2d2f8b2456c74e04b","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-02-13 17:57:37.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"cb33035f415eaef14226b0d20b657e525dca3b31","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2018-02-13 19:02:18.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)\n\nThanks vdrok!","accounts_in_message":[],"_revision_number":8},{"id":"9b6767384fcd290e58ff4b862b713aec1be797df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-13 19:29:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/8/check/openstack-tox-pep8/2db8f4c/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/8/check/openstack-tox-py27/6d01adf/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/8/check/openstack-tox-py35/543ab1a/ : SUCCESS in 4m 19s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/8/check/openstack-tox-pypy/73b77c7/ : SUCCESS in 5m 20s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/8/check/legacy-tempest-dsvm-neutron-src/f951fd7/ : SUCCESS in 1h 30m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/8/check/build-openstack-sphinx-docs/feffb2c/html/ : SUCCESS in 4m 13s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/8/check/build-openstack-releasenotes/8098f87/html/ : SUCCESS in 3m 38s\n- osc-functional-devstack http://logs.openstack.org/99/512699/8/check/osc-functional-devstack/e534ba3/ : SUCCESS in 35m 03s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/8/check/osc-functional-devstack-tips/2f10139/ : FAILURE in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"84c62a4b52a86aecf4f172598e7ad03969435cfb","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-02-13 19:38:32.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"4ee574fa34f46fa93ed1276771f3f1a0bc6710ce","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-02-13 19:38:42.000000000","message":"Patch Set 9:\n\nThanks for review Ruby, done","accounts_in_message":[],"_revision_number":9},{"id":"abeb241308772b9e579de295008252ba6d80801b","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2018-02-13 19:41:27.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)\n\nThanks! I trust that the reno renders properly :)","accounts_in_message":[],"_revision_number":9},{"id":"b5ab9498bacce0fffb2cd4b539fda4ff283615e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-13 21:00:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/9/check/openstack-tox-pep8/ce8a73d/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/9/check/openstack-tox-py27/12d6dbf/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/9/check/openstack-tox-py35/7becbcb/ : SUCCESS in 3m 58s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/9/check/openstack-tox-pypy/556eac9/ : SUCCESS in 5m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/9/check/legacy-tempest-dsvm-neutron-src/2d831d9/ : SUCCESS in 1h 10m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/9/check/build-openstack-sphinx-docs/8567e0c/html/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/9/check/build-openstack-releasenotes/e991442/html/ : SUCCESS in 3m 18s\n- osc-functional-devstack http://logs.openstack.org/99/512699/9/check/osc-functional-devstack/f0c4ec8/ : SUCCESS in 41m 58s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/9/check/osc-functional-devstack-tips/2b7652a/ : SUCCESS in 46m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0dea3f50ba054ebf6a6bfe026117f10edb9fe4ad","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2018-02-16 18:13:27.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"89701d2f723f73a7af4d43dad1e738ef511e662f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2018-02-16 18:14:04.000000000","message":"Patch Set 9: Code-Review-1\n\nThe change itself looks good. Could you bump the minimum keystoneauth in g-r first?","accounts_in_message":[],"_revision_number":9},{"id":"fdcf95004e30c2e5977908d9875f0cd509da8087","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-02-16 18:29:42.000000000","message":"Patch Set 9:\n\nThanks for review Akihiro, will do today","accounts_in_message":[],"_revision_number":9},{"id":"d89e6f0d74c5dacbcacdc138b2e6e4f7fc0ad4e1","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-02-16 18:57:29.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"f1af72b180221e0e18fe665cb728e6b1bdf091b6","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-02-16 18:57:53.000000000","message":"Patch Set 10:\n\nMinimum version bump https://review.openstack.org/545436","accounts_in_message":[],"_revision_number":10},{"id":"dcecad78cfaee182d3e3dd15d6a7fff6670e8c79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-16 20:44:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/10/check/openstack-tox-pep8/f6cf8ae/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/10/check/openstack-tox-py27/980651a/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/10/check/openstack-tox-py35/3cf0988/ : SUCCESS in 4m 57s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/10/check/openstack-tox-pypy/1b229cd/ : SUCCESS in 4m 29s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/10/check/legacy-tempest-dsvm-neutron-src/8973193/ : SUCCESS in 1h 40m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/10/check/build-openstack-sphinx-docs/3b4e92b/html/ : SUCCESS in 3m 19s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/10/check/build-openstack-releasenotes/47542e6/html/ : SUCCESS in 3m 27s\n- osc-functional-devstack http://logs.openstack.org/99/512699/10/check/osc-functional-devstack/59b1658/ : SUCCESS in 44m 56s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/10/check/osc-functional-devstack-tips/1612ff4/ : SUCCESS in 38m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d14603d8a5f71a0dd264e3ce2580a4f941f8d23f","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2018-02-20 16:53:33.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"d175e82d2bd0914be7c2ca0a58a5b6e1bbf50571","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2018-03-13 20:03:12.000000000","message":"Patch Set 10:\n\nrecheck this is fine, need to make sure all is still well after the great functional test strike of 2018","accounts_in_message":[],"_revision_number":10},{"id":"059ea821bddcd5ba209167b17355114e4b85c84b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-13 22:08:02.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/10/check/openstack-tox-pep8/a05b0b6/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/10/check/openstack-tox-py27/f2b446f/ : FAILURE in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/10/check/openstack-tox-py35/b252b21/ : FAILURE in 3m 54s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/10/check/openstack-tox-pypy/297269a/ : FAILURE in 4m 47s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/10/check/legacy-tempest-dsvm-neutron-src/96f8b62/ : FAILURE in 1h 03m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/10/check/build-openstack-sphinx-docs/a9081c2/html/ : SUCCESS in 3m 40s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/10/check/build-openstack-releasenotes/fbde521/html/ : SUCCESS in 3m 22s\n- osc-functional-devstack http://logs.openstack.org/99/512699/10/check/osc-functional-devstack/28f4e43/ : SUCCESS in 1h 07m 24s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/10/check/osc-functional-devstack-tips/695e305/ : SUCCESS in 47m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fc80951a001da291d02414618fd9065210fbb34a","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2018-03-14 07:28:18.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"0c07c0c6816ad997bced65c1741ae312aca6a5b4","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2018-03-14 07:29:16.000000000","message":"Patch Set 11: Code-Review+1\n\nfailures seem to be unrelated to this very change but to some changes in glance API handling that happened in between. Thus trying to rebase.","accounts_in_message":[],"_revision_number":11},{"id":"eb7cb6eefac6d8eda5e6215dfed923e565851a0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 08:32:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/11/check/openstack-tox-pep8/9182d09/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/11/check/openstack-tox-py27/e41bd2d/ : SUCCESS in 7m 11s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/11/check/openstack-tox-py35/8583a35/ : SUCCESS in 6m 52s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/11/check/openstack-tox-pypy/e63b269/ : SUCCESS in 7m 16s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/11/check/legacy-tempest-dsvm-neutron-src/f004360/ : FAILURE in 1h 02m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/11/check/build-openstack-sphinx-docs/2727959/html/ : SUCCESS in 6m 47s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/11/check/build-openstack-releasenotes/0631af9/html/ : SUCCESS in 3m 17s\n- osc-functional-devstack http://logs.openstack.org/99/512699/11/check/osc-functional-devstack/a8eaf9f/ : SUCCESS in 55m 28s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/11/check/osc-functional-devstack-tips/d3f90bc/ : FAILURE in 51m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ec7a957e2ed7fb60adcd9799fe55a4cb11ff845d","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2018-03-14 16:15:28.000000000","message":"Patch Set 11: Code-Review+2\n\nrecheck both failures appear to be transient","accounts_in_message":[],"_revision_number":11},{"id":"1437323abc8049ad140bbccec5bad37a35378a29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 18:18:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/11/check/openstack-tox-pep8/a2a7bba/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/11/check/openstack-tox-py27/8303d09/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/11/check/openstack-tox-py35/9df3f1a/ : SUCCESS in 4m 20s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/11/check/openstack-tox-pypy/df86933/ : SUCCESS in 5m 00s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/11/check/legacy-tempest-dsvm-neutron-src/98b1c44/ : SUCCESS in 1h 34m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/11/check/build-openstack-sphinx-docs/f6d3c3d/html/ : SUCCESS in 3m 56s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/11/check/build-openstack-releasenotes/7f4ffcf/html/ : SUCCESS in 3m 34s\n- osc-functional-devstack http://logs.openstack.org/99/512699/11/check/osc-functional-devstack/f95d223/ : SUCCESS in 47m 29s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/11/check/osc-functional-devstack-tips/4df63a9/ : FAILURE in 49m 10s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d485c7b0a7860751b7bf21073d6002990ac7a46a","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2018-03-14 20:44:33.000000000","message":"Patch Set 11: Workflow+1\n\nThe -tips failure is a common periodic failure and unrelated.  But -tips needs to become voting Real Soon Now","accounts_in_message":[],"_revision_number":11},{"id":"cc5af74aef3e2ab8ffa0eee6ee9990867543d1ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 20:44:45.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"ff6395e94dc81d9b8871280d9fa6940b2d7402e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 22:56:30.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/11/gate/openstack-tox-pep8/c14b8de/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/11/gate/openstack-tox-py27/520b308/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/11/gate/openstack-tox-py35/4ad5010/ : SUCCESS in 3m 54s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/11/gate/legacy-tempest-dsvm-neutron-src/d14d567/ : SUCCESS in 1h 14m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/11/gate/build-openstack-sphinx-docs/2feb88d/html/ : SUCCESS in 3m 51s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/11/gate/build-openstack-releasenotes/c206232/html/ : SUCCESS in 4m 43s\n- osc-functional-devstack http://logs.openstack.org/99/512699/11/gate/osc-functional-devstack/7075122/ : FAILURE in 38m 18s","accounts_in_message":[],"_revision_number":11},{"id":"c060e0fbd79d8abea73079bfae2dcaa0470da2af","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2018-03-15 08:00:10.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"ef7a0ef031d089c9b2349d71935837bc90c533b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 09:40:18.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/11/check/openstack-tox-pep8/7710671/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/11/check/openstack-tox-py27/df7268d/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/11/check/openstack-tox-py35/6c235ec/ : SUCCESS in 4m 02s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/11/check/openstack-tox-pypy/35bf797/ : SUCCESS in 5m 59s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/11/check/legacy-tempest-dsvm-neutron-src/90dad05/ : SUCCESS in 1h 28m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/11/check/build-openstack-sphinx-docs/4cb5483/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/11/check/build-openstack-releasenotes/eb29e79/html/ : SUCCESS in 3m 24s\n- osc-functional-devstack http://logs.openstack.org/99/512699/11/check/osc-functional-devstack/c728b2d/ : POST_FAILURE in 36m 51s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/11/check/osc-functional-devstack-tips/04e79b3/ : POST_FAILURE in 47m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"29ea4f25b6daf22540bd3d139c38991c3dc1be9e","author":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"date":"2018-03-15 09:45:11.000000000","message":"Patch Set 11:\n\nrecheck\n\nthank you for keeping eye on this folks","accounts_in_message":[],"_revision_number":11},{"id":"9a06e4ceff44f6f122e1315e14b4576349e65e58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 11:21:12.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/11/check/openstack-tox-pep8/3ebac5e/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/11/check/openstack-tox-py27/4c9ac1c/ : SUCCESS in 3m 54s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/11/check/openstack-tox-py35/204ef82/ : SUCCESS in 4m 02s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/11/check/openstack-tox-pypy/6a233e5/ : SUCCESS in 4m 38s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/11/check/legacy-tempest-dsvm-neutron-src/f312c9f/ : SUCCESS in 1h 27m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/11/check/build-openstack-sphinx-docs/14cfc12/html/ : SUCCESS in 3m 48s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/11/check/build-openstack-releasenotes/15d50dc/html/ : SUCCESS in 3m 39s\n- osc-functional-devstack http://logs.openstack.org/99/512699/11/check/osc-functional-devstack/10de2d8/ : POST_FAILURE in 43m 45s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/11/check/osc-functional-devstack-tips/0c6f22d/ : POST_FAILURE in 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1492c0772d679e97894cc15195596f747b52b167","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2018-03-15 13:18:17.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"b0884d5d0deff574d1964646e0ea984b441139b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2018-03-15 13:46:47.000000000","message":"Patch Set 11: Code-Review+1\n\nIn combination with https://review.openstack.org/#/c/550573/2, this appears to resolve https://bugs.launchpad.net/bifrost/+bug/1754070.","accounts_in_message":[],"_revision_number":11},{"id":"f464773f855b173846b6bdf18f816faf7269cbfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 14:54:27.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/11/check/openstack-tox-pep8/148c8b6/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/11/check/openstack-tox-py27/982ae88/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/11/check/openstack-tox-py35/b770d6a/ : SUCCESS in 4m 06s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/11/check/openstack-tox-pypy/f975601/ : SUCCESS in 4m 25s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/11/check/legacy-tempest-dsvm-neutron-src/bc2158d/ : SUCCESS in 1h 29m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/11/check/build-openstack-sphinx-docs/6d2ba3d/html/ : SUCCESS in 4m 17s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/11/check/build-openstack-releasenotes/07cf168/html/ : SUCCESS in 4m 16s\n- osc-functional-devstack http://logs.openstack.org/99/512699/11/check/osc-functional-devstack/a8b890b/ : POST_FAILURE in 35m 00s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/11/check/osc-functional-devstack-tips/9455918/ : POST_FAILURE in 41m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"98bef1a336c94035c912f4f5bb0622996dcc417d","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2018-03-15 19:17:03.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"9e4eff0661d33c5dbcd59e0db418481f0e882d28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 22:03:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/11/check/openstack-tox-pep8/b22350e/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/11/check/openstack-tox-py27/7364254/ : SUCCESS in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/11/check/openstack-tox-py35/8473d80/ : SUCCESS in 4m 28s\n- openstack-tox-pypy http://logs.openstack.org/99/512699/11/check/openstack-tox-pypy/9d6673d/ : SUCCESS in 4m 30s (non-voting)\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/11/check/legacy-tempest-dsvm-neutron-src/19f222e/ : SUCCESS in 1h 28m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/11/check/build-openstack-sphinx-docs/d2e4989/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/11/check/build-openstack-releasenotes/e3e90a8/html/ : SUCCESS in 3m 33s\n- osc-functional-devstack http://logs.openstack.org/99/512699/11/check/osc-functional-devstack/f9b878c/ : SUCCESS in 40m 37s\n- osc-functional-devstack-tips http://logs.openstack.org/99/512699/11/check/osc-functional-devstack-tips/4900f12/ : SUCCESS in 44m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"43da8fbbf41b687c5fc88839de093368433cc73c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 22:03:28.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"79073e4ae6d5b727272ebcba494e511831231dac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 23:32:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"0e8d5db3dfa4be1a962d8790e2b661c284904b29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 23:32:51.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/512699/11/gate/openstack-tox-pep8/969f9e2/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/99/512699/11/gate/openstack-tox-py27/33989bb/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/99/512699/11/gate/openstack-tox-py35/32b2fff/ : SUCCESS in 4m 02s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/99/512699/11/gate/legacy-tempest-dsvm-neutron-src/64845f9/ : SUCCESS in 1h 22m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/99/512699/11/gate/build-openstack-sphinx-docs/d256868/html/ : SUCCESS in 3m 54s\n- build-openstack-releasenotes http://logs.openstack.org/99/512699/11/gate/build-openstack-releasenotes/bcaefcb/html/ : SUCCESS in 5m 00s\n- osc-functional-devstack http://logs.openstack.org/99/512699/11/gate/osc-functional-devstack/09a3c69/ : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"b764efc46fcee48394761620672a3a1e117aa3db","revisions":{"b972347ba80484fca7f66656a82a9ad63b34372e":{"kind":"REWORK","_number":1,"created":"2017-10-17 16:26:55.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/1"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:26:41.000000000","tz":180},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/b972347ba80484fca7f66656a82a9ad63b34372e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/b972347ba80484fca7f66656a82a9ad63b34372e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a1b7786211750644c7c24ae52e34e1e9323acc1e":{"kind":"REWORK","_number":2,"created":"2017-10-17 18:20:08.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/2"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 18:19:35.000000000","tz":180},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/a1b7786211750644c7c24ae52e34e1e9323acc1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/a1b7786211750644c7c24ae52e34e1e9323acc1e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fcbab0f9f30f329d8d32f904f0a09a33196418a3":{"kind":"REWORK","_number":3,"created":"2017-10-17 18:32:32.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/3"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 18:30:23.000000000","tz":0},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/fcbab0f9f30f329d8d32f904f0a09a33196418a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/fcbab0f9f30f329d8d32f904f0a09a33196418a3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"96a2a5e61695df764d08a1dc8712b9f020db0cf2":{"kind":"REWORK","_number":4,"created":"2017-10-27 14:44:09.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/4"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-27 12:18:42.000000000","tz":180},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/96a2a5e61695df764d08a1dc8712b9f020db0cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/96a2a5e61695df764d08a1dc8712b9f020db0cf2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"46b1d6c5a665f6dffc28619f3b3f40bed8c22f86":{"kind":"REWORK","_number":5,"created":"2017-10-27 15:00:00.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/5"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-27 14:59:56.000000000","tz":0},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/46b1d6c5a665f6dffc28619f3b3f40bed8c22f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/46b1d6c5a665f6dffc28619f3b3f40bed8c22f86"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2858b41d95d7251afcc523ce66988ad91abb0974":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2017-10-27 15:02:28.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/6"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-27 15:02:25.000000000","tz":0},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nDepends-On: I3e477895ba0c989ffd0c91c45791e9f74173a3d6\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/2858b41d95d7251afcc523ce66988ad91abb0974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/2858b41d95d7251afcc523ce66988ad91abb0974"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fb8398e1cb94833601a06ef71f4baf5e4a72d413":{"kind":"REWORK","_number":7,"created":"2018-02-01 02:02:22.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/7"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2018-02-01 02:02:19.000000000","tz":0},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nDepends-On: I3e477895ba0c989ffd0c91c45791e9f74173a3d6\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/fb8398e1cb94833601a06ef71f4baf5e4a72d413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/fb8398e1cb94833601a06ef71f4baf5e4a72d413"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"937373ce86798726ef1239df76d095ac9b8f7814":{"kind":"REWORK","_number":8,"created":"2018-02-13 17:57:37.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/8"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2018-02-13 17:57:32.000000000","tz":0},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nDepends-On: I3e477895ba0c989ffd0c91c45791e9f74173a3d6\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/937373ce86798726ef1239df76d095ac9b8f7814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/937373ce86798726ef1239df76d095ac9b8f7814"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f0f9349c807876fb30e95d6165d340eac719f67b":{"kind":"REWORK","_number":9,"created":"2018-02-13 19:38:32.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/9"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2018-02-13 19:38:25.000000000","tz":0},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nDepends-On: I3e477895ba0c989ffd0c91c45791e9f74173a3d6\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/f0f9349c807876fb30e95d6165d340eac719f67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/f0f9349c807876fb30e95d6165d340eac719f67b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ab7a3986877d18c90a9cf8e873cbf1b2c1891377":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2018-02-16 18:57:29.000000000","uploader":{"_account_id":12356,"name":"Vladyslav Drok","email":"vdrok@mirantis.com","username":"vdrok"},"ref":"refs/changes/99/512699/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/10"}}},"commit":{"parents":[{"commit":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/5de9eb4406cd799e796c68746f3adc3ee9b9f6ed"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2018-02-16 18:57:26.000000000","tz":0},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nDepends-On: I3e477895ba0c989ffd0c91c45791e9f74173a3d6\nDepends-On: I3de32193f41d9d1012043b43ae8080f3b1e828e5\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/ab7a3986877d18c90a9cf8e873cbf1b2c1891377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/ab7a3986877d18c90a9cf8e873cbf1b2c1891377"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5de9eb4406cd799e796c68746f3adc3ee9b9f6ed","is_merged_in_target_branch":true,"change_id":"Idcbd354b956d307cfaa0a85ce6b413773d7221fb","change_number":509448,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b764efc46fcee48394761620672a3a1e117aa3db":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2018-03-14 07:28:18.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/99/512699/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/99/512699/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/99/512699/11"}}},"commit":{"parents":[{"commit":"22de9c2bbfbbdea9adbfa29c1a056e11fe947e25","subject":"Set defaults in ClientManager tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/22de9c2bbfbbdea9adbfa29c1a056e11fe947e25"}]}],"author":{"name":"Vladyslav Drok","email":"vdrok@mirantis.com","date":"2017-10-17 16:21:53.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","date":"2018-03-14 07:28:18.000000000","tz":0},"subject":"Allow to use the none auth plugin","message":"Allow to use the none auth plugin\n\nThis change allows to use the none auth plugin in CLI without having\nto specify --os-cloud file with basically a single endpoint parameter.\n\nCloses-Bug: #1724283\nDepends-On: I3e477895ba0c989ffd0c91c45791e9f74173a3d6\nDepends-On: I3de32193f41d9d1012043b43ae8080f3b1e828e5\nChange-Id: I6cf4e22d676f4a55f84863125aee7318915a7404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/b764efc46fcee48394761620672a3a1e117aa3db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/b764efc46fcee48394761620672a3a1e117aa3db"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"22de9c2bbfbbdea9adbfa29c1a056e11fe947e25","is_merged_in_target_branch":true,"change_id":"I48dd3762377dd8b17ce039acba8c61116c798734","change_number":552716,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
