)]}'
{"id":"openstack%2Fosc-lib~724904","triplet_id":"openstack%2Fosc-lib~master~Id71c701b8053af5f8ab26c9f6fa66140e76aa8fc","project":"openstack/osc-lib","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"Id71c701b8053af5f8ab26c9f6fa66140e76aa8fc","subject":"Use mock from unittest","status":"MERGED","created":"2020-05-01 18:28:03.000000000","updated":"2020-05-06 17:45:33.000000000","submitted":"2020-05-06 17:43:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724904-1588787023304-a9cd72ef","meta_rev_id":"b8e4786cb8f2b6ae8c43eca0815cc5994ce15fe0","_number":724904,"virtual_id_number":724904,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-05-02 09:58:46.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 17:43:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-01 18:39:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-06 15:55:45.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-06 15:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-06 15:55:45.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-01 18:39:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-06 15:30:32.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-05-06 15:55:45.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-06 17:43:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc0a78f65ce47ace6eee1824473be95fbba77da2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-01 18:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98edf6f5d46d60e168aafa4f884267f17b0757bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-01 18:37:30.000000000","message":"Topic changed from unittest-mock to unittest.mock","accounts_in_message":[],"_revision_number":1},{"id":"e9744e78f42200e7dc840c39f206e36983dfe29d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-01 18:39:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nJust noting some potential need for import ordering cleanup, but that\u0027s outside the scope of this patch. Just pointing out for awareness.","accounts_in_message":[],"_revision_number":1},{"id":"a3ebb140ae87eeff4a29e46ce2c9cd91ae19b687","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-01 18:45:17.000000000","message":"Patch Set 1:\n\n\u003e (3 comments)\n \u003e \n \u003e Just noting some potential need for import ordering cleanup, but\n \u003e that\u0027s outside the scope of this patch. Just pointing out for\n \u003e awareness.\n\nHey Sean, I\u0027m aware of the wrong import ordering and I\u0027m doing a separate patch to enforce that, I can rebase that on top of this or vice-versa.","accounts_in_message":[],"_revision_number":1},{"id":"c17f0c3b3458959cc8ee6e223202f3170730a4af","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-01 18:48:13.000000000","message":"Patch Set 1:\n\n\u003e \u003e (3 comments)\n \u003e \u003e\n \u003e \u003e Just noting some potential need for import ordering cleanup, but\n \u003e \u003e that\u0027s outside the scope of this patch. Just pointing out for\n \u003e \u003e awareness.\n \u003e \n \u003e Hey Sean, I\u0027m aware of the wrong import ordering and I\u0027m doing a\n \u003e separate patch to enforce that, I can rebase that on top of this or\n \u003e vice-versa.\n\nAwesome. Whatever is easy for you. I\u0027m not saying you need to be the one to do it either. I just wanted it pointed out to raise awareness that it is something that should probably be cleaned up at some point.","accounts_in_message":[],"_revision_number":1},{"id":"3a43bb97be45ce63a397d9a0bc002b3e7fc965ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 19:49:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7cb88a493ed54608abd8452a9369a69b : SUCCESS in 4m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adf37a678f89423c9d06b215f56825d9 : POST_FAILURE in 1h 12m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbe187ce360f4cad80b1c7f382778eda : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a278b64d921c4d1e9d9bb573f26bf27b : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e60fd8acb449bcb6b224341757276b : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b838166e67a4aeca1309af2c61e23be : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e94d093e13c48b4a611d94ea4d8c970 : SUCCESS in 4m 43s (non-voting)\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/4e93322f79b24799aa2cee360440d44b : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed083f5964a4bd0b9c0a3769f435636 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb7d9070ac0d44baa251800533c4d0a3 : SUCCESS in 4m 39s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/382e5732dfe74968b4a508e8922ca446 : POST_FAILURE in 1h 00m 22s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3a76691e85dd4faeb582d30fab9f19a2 : POST_FAILURE in 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8c103997f52492a2c4d5223064d4dfca84f60e7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-02 09:58:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b387784383979a0ef7a775611ba59d55632b81e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 11:24:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8371979010ca42e9a818d9bd72dc22c2 : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f40d5ef9c9c548f1a7f39da59a402383 : SUCCESS in 1h 20m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4d287865fc44b2f9b2a11e101e7d8cc : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/976e5a5d55a94b78aa122e4e988f140d : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6af7382b6ecc4d3f97e9a986f18d6176 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6937714b90e94a24bf1eff34ec5af9b5 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaccf76d32e5429f9a8b225295a863b5 : SUCCESS in 3m 37s (non-voting)\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/49b8f841407448c69494ff686f75c8dd : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c360f4df73b48ce81d22a97d5d3c5f3 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33332090bd12478f88dade66ad30e8cd : SUCCESS in 3m 16s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/664a2b977cd240539ed41e871d196ee1 : SUCCESS in 50m 50s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4b2e040f78754ccba7eab8c15f701195 : SUCCESS in 55m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4eb6e86fefdb5d1e6ef52d32456909043c2dbb9a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-06 15:30:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c12ffad710cc9ca7bf2e995bc4ecd062d2edf9e3","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-06 15:55:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ccb99e040d74c8f5a85df41a87e335cc13991d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 15:55:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d10cc58d8315b4d72aa60d49d07e006c65231c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 17:43:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cc1652436564889a97cbdfb3b08cad8 : SUCCESS in 3m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d0eee66e73441558945a06646e2657f : SUCCESS in 1h 44m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a47eb279ecfa4b99a03c5b8140348cba : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfcdff6a5e764a60b7e117913db337d3 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5917eed74d654b4c88f3cf5f53604a59 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c24b0d4752b4b00a269fb0dd1f86120 : SUCCESS in 4m 50s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/ae84f48bfd3d43fa82ed780b96153560 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a1705370d145a4a415405a0920e882 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78b3b7f50bb649cb9d330e58d4bcd4cd : SUCCESS in 3m 53s\n- osc-functional-devstack https://zuul.opendev.org/t/openstack/build/bdeb454e1371443cbc43eb1fe72978b3 : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0a2fcbde06dd2a807e7bead9a7a38dfb4d210675","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 17:43:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8e4786cb8f2b6ae8c43eca0815cc5994ce15fe0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 17:45:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c95b4d6fad4137a62f551d709567b7 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9ae997b336b4fac9d7a274ce3f65e95 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d98486d0619c0729f3962d9c0db50d9ef156a50","revisions":{"2d98486d0619c0729f3962d9c0db50d9ef156a50":{"kind":"REWORK","_number":1,"created":"2020-05-01 18:28:03.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/04/724904/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-lib","ref":"refs/changes/04/724904/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/04/724904/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/04/724904/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-lib refs/changes/04/724904/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-lib refs/changes/04/724904/1"}}},"commit":{"parents":[{"commit":"b7f68464809f00c2fb7dd311d003c7e71524e819","subject":"Merge \"Remove Babel from requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/b7f68464809f00c2fb7dd311d003c7e71524e819"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-01 18:20:29.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-01 18:28:01.000000000","tz":120},"subject":"Use mock from unittest","message":"Use mock from unittest\n\nSince mock is part of unittest std lib now, we can just use it and\nremove the external dependency.\n\nChange-Id: Id71c701b8053af5f8ab26c9f6fa66140e76aa8fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/2d98486d0619c0729f3962d9c0db50d9ef156a50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-lib/commit/2d98486d0619c0729f3962d9c0db50d9ef156a50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
