)]}'
{"id":"openstack%2Fosc-placement~1004597","triplet_id":"openstack%2Fosc-placement~stable%2F2026.2~I5da3cdb2afac3d1eaf2c620c4e699cd46e3206d3","project":"openstack/osc-placement","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"I5da3cdb2afac3d1eaf2c620c4e699cd46e3206d3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","status":"MERGED","created":"2026-09-08 13:26:34.000000000","updated":"2026-09-11 13:53:47.000000000","submitted":"2026-09-11 13:49:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004597-create-2026.2","meta_rev_id":"08775702bd9328d6cdf34d9923fe70fc9bede277","_number":1004597,"virtual_id_number":1004597,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 13:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2026-09-11 13:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":1,"date":"2026-09-11 13:36:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 13:43:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 13:36:24.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"2f183c1826d078d8c7ef4c6b449fa0aa62f674f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-08 13:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a168ecb92e8e9fd6a4364979d7e5ccaef40f6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 13:43:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5786d43e86ca4132a61100476e79080f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa307a4c776403484a9fdf01e30f6f4 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50ccfa4e8d794b31845eae70c9564b74 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59f009d3dca54c4ba2bea77cd35ad3b5 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b798c85460014f25a228a1e9b93d4b1b : SUCCESS in 6m 13s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/87a41cc154ac438a875261fc548b8b17 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f117c292e2844f6594a8c7e4b8066279 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae3d0dc6519741bd82d2a4f65217dc9d : SUCCESS in 2m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3faf5af61d3b42379254fb6f1addc717 : SUCCESS in 2m 48s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f5e45a75fd074d5392d3d297a3f83bc4 : SUCCESS in 4m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/06371df447204c26a01056ec9623985e : SUCCESS in 2m 45s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/2d3890b187da47168ecc9e139eec73d8 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"81b8ceae9d764060964b7063531f7676338947a2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-11 13:36:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ae11d0432d019badd0f20fe4a654e43ec49c17a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:36:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e14f8e11e047de424dc3fcff08adb23a5a5e8004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:49:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9787b41fb0c84be09b85fb20e667983b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc97232df1142339bff79a52e9d3521 : SUCCESS in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4771aee51d0845a6a82dfa54ccb186a4 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c15677f64c3149e0ad04afb3f943f8e7 : SUCCESS in 2m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db06da33d0784268af49a6622618c0a9 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65d0d5fcde84b44b1968adf5525e313 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/658654db4589492499aff1065914f555 : SUCCESS in 2m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/86cbb1037ce7434e877bd725c572745c : SUCCESS in 4m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1ae0e5d8e943473d8952f049bc857766 : SUCCESS in 4m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/01c4b4fdd2d442d3a13f22fec34dd44f : SUCCESS in 4m 08s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6aa3ae2f2ff747e9a807863404284c61 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4de074d78067ab8bc9bd498234a536bc620455f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:49:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08775702bd9328d6cdf34d9923fe70fc9bede277","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:53:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63836c7cb36141eab7403c23ae86d78e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be138a6f154d4fa78729149babd5a106 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04566ff2479f4f1b87667cf40c2988b7 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"402610a218722703381079a3e89935c92dc6f46b","revisions":{"402610a218722703381079a3e89935c92dc6f46b":{"kind":"REWORK","_number":1,"created":"2026-09-08 13:26:34.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/97/1004597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/97/1004597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/97/1004597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/97/1004597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/97/1004597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/97/1004597/1"}}},"commit":{"parents":[{"commit":"61499f1191214ef69dee17e4b549dd470b3ea746","subject":"Update .gitreview for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/61499f1191214ef69dee17e4b549dd470b3ea746"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-08 13:26:33.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-08 13:26:33.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I5da3cdb2afac3d1eaf2c620c4e699cd46e3206d3\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/402610a218722703381079a3e89935c92dc6f46b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/402610a218722703381079a3e89935c92dc6f46b"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
