)]}'
{"id":"openstack%2Fosc-placement~651783","triplet_id":"openstack%2Fosc-placement~master~I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e","project":"openstack/osc-placement","branch":"master","hashtags":[],"change_id":"I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e","subject":"Add support for 1.22 microversion","status":"MERGED","created":"2019-04-11 13:20:19.000000000","updated":"2019-06-03 12:48:34.000000000","submitted":"2019-06-03 12:48:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651783-1559566114411-e548dae4","meta_rev_id":"b118a910e2b50b79f54bb895697dc2d82b9508b0","_number":651783,"virtual_id_number":651783,"owner":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":2,"date":"2019-06-03 12:48:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"},{"value":0,"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"recommended":{"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},"all":[{"value":1,"date":"2019-05-31 08:45:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-05-31 08:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-31 08:49:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-06-03 12:19:29.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2019-05-31 08:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"},{"value":0,"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2019-06-03 12:19:29.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"},{"value":0,"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"},{"value":0,"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},{"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},{"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 20:10:00.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2019-05-24 15:50:11.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-05-28 05:58:44.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-05-31 08:45:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-05-31 08:49:38.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-05-31 08:53:59.000000000","updated_by":{"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"},"reviewer":{"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"},"state":"REVIEWER"},{"updated":"2019-05-31 08:56:44.000000000","updated_by":{"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},"reviewer":{"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},"state":"REVIEWER"},{"updated":"2019-06-03 12:19:29.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-03 12:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bacce10330ae85111a4be9f708c71c67f58752f4","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-04-11 13:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7d86fd7d2965cdfdcc107520c2426fddd620f98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 13:41:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/83/651783/1/check/openstack-tox-pep8/f798587/ : FAILURE in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/83/651783/1/check/openstack-tox-py27/5d2db2b/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/83/651783/1/check/openstack-tox-py35/47128b3/ : SUCCESS in 6m 49s\n- openstack-tox-py36 http://logs.openstack.org/83/651783/1/check/openstack-tox-py36/226dcbe/ : SUCCESS in 4m 48s\n- openstack-tox-py37 http://logs.openstack.org/83/651783/1/check/openstack-tox-py37/b8f5cfd/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/83/651783/1/check/openstack-tox-docs/bfb41d0/html/ : SUCCESS in 4m 32s\n- build-openstack-releasenotes http://logs.openstack.org/83/651783/1/check/build-openstack-releasenotes/d99dc6e/html/ : SUCCESS in 5m 53s\n- osc-placement-dsvm-functional http://logs.openstack.org/83/651783/1/check/osc-placement-dsvm-functional/1e62715/ : SUCCESS in 14m 06s\n- osc-placement-dsvm-functional-py3 http://logs.openstack.org/83/651783/1/check/osc-placement-dsvm-functional-py3/b6061a3/ : SUCCESS in 18m 28s","accounts_in_message":[],"_revision_number":1},{"id":"27f8fc1df83339a8f0ab7494362b8d3313fbc5c9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-11 13:43:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nThere are some missing test cases but the code looks good. Thanks for picking this task up.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":1},{"id":"3da995c77bfcfa711752f18270f1b28007d4f227","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-04-15 11:27:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40fa59e8924c696b3fdc8cfe54dc7e1680ce28fb","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-04-15 11:30:08.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6745b24a120930bc0b69e20320f231d8471c7b79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 11:57:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/83/651783/2/check/openstack-tox-pep8/eabd011/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/83/651783/2/check/openstack-tox-py27/2338dc2/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/83/651783/2/check/openstack-tox-py35/791235f/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/83/651783/2/check/openstack-tox-py36/81cd29e/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/83/651783/2/check/openstack-tox-py37/6784b09/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/83/651783/2/check/openstack-tox-docs/cafdd47/html/ : SUCCESS in 4m 23s\n- build-openstack-releasenotes http://logs.openstack.org/83/651783/2/check/build-openstack-releasenotes/95a82ae/html/ : SUCCESS in 4m 13s\n- osc-placement-dsvm-functional http://logs.openstack.org/83/651783/2/check/osc-placement-dsvm-functional/26ee380/ : SUCCESS in 20m 02s\n- osc-placement-dsvm-functional-py3 http://logs.openstack.org/83/651783/2/check/osc-placement-dsvm-functional-py3/cc55515/ : SUCCESS in 23m 08s","accounts_in_message":[],"_revision_number":2},{"id":"d6b2e5bc06582f80da76f78f25defed0692b2ada","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-15 12:07:33.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for doing this, looks really good.\n\nI think we should get a few more reviews to make sure that everyone agrees with the syntax and the docs, so I\u0027ll just +1 for now.","accounts_in_message":[],"_revision_number":2},{"id":"887a758e2bfc7ba1ff6eee363ed977bfdd5e28d3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-04-15 12:48:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"107c5ffd93073a1fd965407af8e0847742af0ea2","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-15 15:04:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)\n\nIt works for me though ! had to be escaped.\nSoft -1 on the typo in the release note.","accounts_in_message":[],"_revision_number":2},{"id":"b09810663d5f3fee3d25699a87455ca8a64947e3","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-04-16 08:32:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0db6e29146e52430fd01bc16228079b695d298ed","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-04-16 08:34:21.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nThanks for pointing this out.","accounts_in_message":[],"_revision_number":2},{"id":"4aaf3e93a7e15ebe26b1b96f1ae72e7db9f2eb88","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-16 08:39:08.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks Pawel! LGTM","accounts_in_message":[],"_revision_number":3},{"id":"847f169ad61fc5856102dcead671ff290484e272","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 09:08:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/83/651783/3/check/openstack-tox-pep8/751814c/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/83/651783/3/check/openstack-tox-py27/d552921/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/83/651783/3/check/openstack-tox-py35/5e64f92/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/83/651783/3/check/openstack-tox-py36/1782e7b/ : SUCCESS in 4m 00s\n- openstack-tox-py37 http://logs.openstack.org/83/651783/3/check/openstack-tox-py37/4e75585/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/83/651783/3/check/openstack-tox-docs/7a38e70/html/ : SUCCESS in 4m 00s\n- build-openstack-releasenotes http://logs.openstack.org/83/651783/3/check/build-openstack-releasenotes/cc136af/html/ : SUCCESS in 3m 52s\n- osc-placement-dsvm-functional http://logs.openstack.org/83/651783/3/check/osc-placement-dsvm-functional/cbbd5b4/ : SUCCESS in 22m 39s\n- osc-placement-dsvm-functional-py3 http://logs.openstack.org/83/651783/3/check/osc-placement-dsvm-functional-py3/2f37965/ : SUCCESS in 23m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d52751ec6bd6e13e37eb3d47db4de7c9c23ea838","author":{"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"},"date":"2019-04-16 12:20:57.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good for me :D","accounts_in_message":[],"_revision_number":3},{"id":"b2d54279c86e82de00b5a8e6ee3c9f3976f17786","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-16 20:09:57.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"190ae66a5d5649434468b6b81f52ad4121bedeb5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-16 20:10:13.000000000","message":"Patch Set 3:\n\nMaybe Dean has some experience on another CLI that tries to do something like this.","accounts_in_message":[],"_revision_number":3},{"id":"55c69de589ad789d34c88be3624c562037fcf370","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-16 20:18:52.000000000","message":"Patch Set 3:\n\nMatt\u0027s comments are why I didn\u0027t leave much of a vote before.\n\nI want the client to reflect the way the API expresses things because that\u0027s the way I think: the API is the important thing, the client is a convenience that should operate as an on ramp to understanding the structure, grammar and expressiveness of the API.\n\nThis is not at all what openstack CLI in general is trying to do or be (at least as far as I can tell).\n\nThe bit about forbidden member_of in microversion 1.32 is a compelling (to me) reason to carry on with `!` here. We\u0027d like to avoid a collision later.\n\nBut I\u0027m saying that as\n\na) the guy who decided we should use ! in the query parameters\nb) the guy who said the above about the CLI being an onramp to understanding the API\n\nSo even though I\u0027m personally perfectly fine with the way the client is expressing itself now, I\u0027ve got a big bias that could very well be wrong.","accounts_in_message":[],"_revision_number":3},{"id":"75bf5b2f47c8a2d778fdee8307bd56f55a97b3dd","author":{"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},"date":"2019-04-17 10:32:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8a1d17f3acc8cc30cd7f9816d8c1adffed4009d4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-08 10:08:01.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aff6c3e160258328d7556a40f3320a60caa382cf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-13 15:22:39.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"91819b3b560f6a6507b3802405e3f66c0689eed1","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-05-14 05:41:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6eaa0576920507878ed9661e18b34ebac9037932","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-14 07:53:31.000000000","message":"Patch Set 3:\n\nSounds good to me with required/forbidden for traits and --aggregate-uuid/forbidden-aggregate-uuid for aggregates","accounts_in_message":[],"_revision_number":3},{"id":"74c4f3a6803a443c06d0d308738a29db8c1cadff","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-24 15:50:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3fb9182915b701cb129aaa366deb1b514bc36616","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-27 08:47:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2cde870e3248fc29ab6ef7658a170ad07a0a1aa2","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-28 05:58:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c2793c02ab48afb1b4c88dea9e03e8ba0225609f","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-05-30 13:46:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1ef1f503b547bd248b4b999bf85672bd74b1d480","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 13:46:13.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"97dcc18a91e44b8e0c9039aeaac9d6fe509d523a","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-05-30 13:51:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nmerge conflict to be fixed","accounts_in_message":[],"_revision_number":4},{"id":"cdf7537643f621ee2e8e791ad67b05af15c14338","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-05-31 06:58:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c49fda7cadb2c5012eee37577d5990401ca1e41e","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-05-31 06:59:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f517aadfa07d234028f697b6d24ed28f51b53a59","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-05-31 07:00:52.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2b4574faef7d23e05bc57da89db9b3a3cac678dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 07:11:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/83/651783/6/check/openstack-tox-pep8/7659a88/ : FAILURE in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/83/651783/6/check/openstack-tox-py27/219b564/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/83/651783/6/check/openstack-tox-py36/56ddd91/ : SUCCESS in 4m 25s\n- openstack-tox-py37 http://logs.openstack.org/83/651783/6/check/openstack-tox-py37/c982f49/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/83/651783/6/check/openstack-tox-docs/60ce65c/html/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes http://logs.openstack.org/83/651783/6/check/build-openstack-releasenotes/f6ffc57/html/ : SUCCESS in 3m 47s\n- openstack-tox-functional http://logs.openstack.org/83/651783/6/check/openstack-tox-functional/f55d6f6/ : SUCCESS in 4m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/651783/6/check/openstack-tox-functional-py36/3c5414d/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":6},{"id":"33127258f6be5194103feea8dd1e2b977ed70bd1","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-05-31 07:47:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4be8fcbefbdffd828a018a66d43f6611dd97496f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 07:57:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/83/651783/7/check/openstack-tox-pep8/cd06208/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/83/651783/7/check/openstack-tox-py27/89f6ec6/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/83/651783/7/check/openstack-tox-py36/c334802/ : SUCCESS in 3m 50s\n- openstack-tox-py37 http://logs.openstack.org/83/651783/7/check/openstack-tox-py37/d843e31/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/83/651783/7/check/openstack-tox-docs/be8735b/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/83/651783/7/check/build-openstack-releasenotes/a3d094b/html/ : SUCCESS in 3m 26s\n- openstack-tox-functional http://logs.openstack.org/83/651783/7/check/openstack-tox-functional/d5c40e6/ : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/651783/7/check/openstack-tox-functional-py36/181038c/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":7},{"id":"71e6d84a9103f8499455667b48bce86fe83c268e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-31 08:45:29.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5c89847b7fe5fc6bcd9c0b85c19296b30e23e761","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-31 08:49:38.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks Pawel! LGTM\nCheers,\ngibi","accounts_in_message":[],"_revision_number":7},{"id":"228fa5251e74f061443bcf7a37946ee6299adb2d","author":{"_account_id":30237,"name":"Filip Pich","email":"filip_pich@o2.pl"},"date":"2019-05-31 08:53:59.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)\n\ngood job","accounts_in_message":[],"_revision_number":7},{"id":"7a31e2d66ad928c6c8a8901333e35446cbd8be45","author":{"_account_id":30202,"name":"Pawel Tucholski","email":"pawel.adam.tucholski@gmail.com","username":"pawel.tucholski"},"date":"2019-05-31 08:56:44.000000000","message":"Patch Set 7: Code-Review+1\n\ngreat","accounts_in_message":[],"_revision_number":7},{"id":"e81caf9c43c6c5077150035a1c76b0e14224d4ee","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-03 12:19:29.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLet\u0027s make this go. I mentioned a nit within, but it\u0027s too much of a nit to hold this up. Thanks for the effort getting this done.","accounts_in_message":[],"_revision_number":7},{"id":"bcb8f21ffe472e799203a2591baabaa3787a3e05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 12:19:40.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"fa9fe45bc134c649596571694f00bbe18bc109df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 12:48:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"b118a910e2b50b79f54bb895697dc2d82b9508b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 12:48:34.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/83/651783/7/gate/openstack-tox-pep8/6679a6c/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/83/651783/7/gate/openstack-tox-py27/637d970/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/83/651783/7/gate/openstack-tox-py36/2c37d67/ : SUCCESS in 4m 57s\n- openstack-tox-py37 http://logs.openstack.org/83/651783/7/gate/openstack-tox-py37/f162ab7/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/83/651783/7/gate/openstack-tox-docs/86622c0/html/ : SUCCESS in 4m 48s\n- build-openstack-releasenotes http://logs.openstack.org/83/651783/7/gate/build-openstack-releasenotes/d32ce00/html/ : SUCCESS in 5m 11s\n- openstack-tox-functional http://logs.openstack.org/83/651783/7/gate/openstack-tox-functional/91aaeec/ : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/83/651783/7/gate/openstack-tox-functional-py36/d4392c9/ : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e52fa969f9ee7ff2e58ce86c2a6261a80b819093","revisions":{"2e12c3f70e1c5711345264e41441f42b5264383f":{"kind":"REWORK","_number":1,"created":"2019-04-11 13:20:19.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/83/651783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/83/651783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/1"}}},"commit":{"parents":[{"commit":"d7c1f33d06e210e718830bef1d2cd3a9377c02ea","subject":"Microversion 1.21 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d7c1f33d06e210e718830bef1d2cd3a9377c02ea"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-02 13:21:21.000000000","tz":0},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-11 13:16:35.000000000","tz":120},"subject":"Add support for 1.22 microversion","message":"Add support for 1.22 microversion\n\nThe 1.22 microversion of placement adds support for\nexcluding resource providers and allocations candidates\nwith specified traits. A forbidden trait (as --required\nargument) should be prefixed with a ! sign.\n\nChange-Id: I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2e12c3f70e1c5711345264e41441f42b5264383f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2e12c3f70e1c5711345264e41441f42b5264383f"}]},"branch":"refs/heads/master"},"0459791e96b045c88ebd0f3711ae6a9b9383134f":{"kind":"REWORK","_number":2,"created":"2019-04-15 11:27:47.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/83/651783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/83/651783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/2"}}},"commit":{"parents":[{"commit":"d7c1f33d06e210e718830bef1d2cd3a9377c02ea","subject":"Microversion 1.21 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d7c1f33d06e210e718830bef1d2cd3a9377c02ea"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-02 13:21:21.000000000","tz":0},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-15 11:25:59.000000000","tz":120},"subject":"Add support for 1.22 microversion","message":"Add support for 1.22 microversion\n\nThe 1.22 microversion of placement adds support for\nexcluding resource providers and allocations candidates\nwith specified traits. A forbidden trait (as --required\nargument) should be prefixed with a ! sign.\n\nChange-Id: I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0459791e96b045c88ebd0f3711ae6a9b9383134f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0459791e96b045c88ebd0f3711ae6a9b9383134f"}]},"branch":"refs/heads/master"},"a809cdd4e72cd3ca44b50582beca48b9dd5be1c4":{"kind":"REWORK","_number":3,"created":"2019-04-16 08:32:09.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/83/651783/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/83/651783/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/3"}}},"commit":{"parents":[{"commit":"d7c1f33d06e210e718830bef1d2cd3a9377c02ea","subject":"Microversion 1.21 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d7c1f33d06e210e718830bef1d2cd3a9377c02ea"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-02 13:21:21.000000000","tz":0},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-16 08:28:35.000000000","tz":120},"subject":"Add support for 1.22 microversion","message":"Add support for 1.22 microversion\n\nThe 1.22 microversion of placement adds support for\nexcluding resource providers and allocation candidates\nwith specified traits. A forbidden trait (as --required\nargument) should be prefixed with a ! sign.\n\nChange-Id: I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a809cdd4e72cd3ca44b50582beca48b9dd5be1c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a809cdd4e72cd3ca44b50582beca48b9dd5be1c4"}]},"branch":"refs/heads/master"},"672d0b1f7e4e265a2fda231016a7c63fe75a30e5":{"kind":"REWORK","_number":4,"created":"2019-05-30 13:46:02.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/83/651783/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/83/651783/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/4"}}},"commit":{"parents":[{"commit":"d7c1f33d06e210e718830bef1d2cd3a9377c02ea","subject":"Microversion 1.21 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d7c1f33d06e210e718830bef1d2cd3a9377c02ea"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-02 13:21:21.000000000","tz":0},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-05-30 13:36:59.000000000","tz":120},"subject":"Add support for 1.22 microversion","message":"Add support for 1.22 microversion\n\nThe 1.22 microversion of placement adds support for\nexcluding resource providers and allocation candidates\nwith specified traits. A forbidden trait may be specified\nwith --forbidden option.\n\nChange-Id: I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/672d0b1f7e4e265a2fda231016a7c63fe75a30e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/672d0b1f7e4e265a2fda231016a7c63fe75a30e5"}]},"branch":"refs/heads/master"},"610a6a0519fe6555667ccb06ace69968c6d97a16":{"kind":"REWORK","_number":5,"created":"2019-05-31 06:58:18.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/83/651783/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/83/651783/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/5"}}},"commit":{"parents":[{"commit":"4c95f7caff9bf4a81afd349577df5615a0f83807","subject":"Merge \"Expose version error message generically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4c95f7caff9bf4a81afd349577df5615a0f83807"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-02 13:21:21.000000000","tz":0},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-05-31 06:54:50.000000000","tz":120},"subject":"Add support for 1.22 microversion","message":"Add support for 1.22 microversion\n\nThe 1.22 microversion of placement adds support for\nexcluding resource providers and allocation candidates\nwith specified traits. A forbidden trait may be specified\nwith --forbidden option.\n\nChange-Id: I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/610a6a0519fe6555667ccb06ace69968c6d97a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/610a6a0519fe6555667ccb06ace69968c6d97a16"}]},"branch":"refs/heads/master"},"e41a1eb9a9b9425b0432fefed5a7e0e06ccf0349":{"kind":"REWORK","_number":6,"created":"2019-05-31 06:59:59.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/83/651783/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/83/651783/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/6"}}},"commit":{"parents":[{"commit":"4c95f7caff9bf4a81afd349577df5615a0f83807","subject":"Merge \"Expose version error message generically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4c95f7caff9bf4a81afd349577df5615a0f83807"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-02 13:21:21.000000000","tz":0},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-05-31 06:59:50.000000000","tz":120},"subject":"Add support for 1.22 microversion","message":"Add support for 1.22 microversion\n\nThe 1.22 microversion of placement adds support for\nexcluding resource providers and allocation candidates\nwith specified traits. A forbidden trait may be specified\nwith --forbidden option.\n\nChange-Id: I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e41a1eb9a9b9425b0432fefed5a7e0e06ccf0349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e41a1eb9a9b9425b0432fefed5a7e0e06ccf0349"}]},"branch":"refs/heads/master"},"e52fa969f9ee7ff2e58ce86c2a6261a80b819093":{"kind":"REWORK","_number":7,"created":"2019-05-31 07:47:20.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/83/651783/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/83/651783/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/83/651783/7"}}},"commit":{"parents":[{"commit":"4c95f7caff9bf4a81afd349577df5615a0f83807","subject":"Merge \"Expose version error message generically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/4c95f7caff9bf4a81afd349577df5615a0f83807"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-04-02 13:21:21.000000000","tz":0},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-05-31 07:47:13.000000000","tz":120},"subject":"Add support for 1.22 microversion","message":"Add support for 1.22 microversion\n\nThe 1.22 microversion of placement adds support for\nexcluding resource providers and allocation candidates\nwith specified traits. A forbidden trait may be specified\nwith --forbidden option.\n\nChange-Id: I4760fe6119e29d80f63ec7bd6c4a5f2395a9291e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
