)]}'
{"id":"openstack%2Fosc-placement~653285","triplet_id":"openstack%2Fosc-placement~master~I28226a1dbc003a8b6366e528469152c141b40593","project":"openstack/osc-placement","branch":"master","topic":"story/2005448","hashtags":[],"change_id":"I28226a1dbc003a8b6366e528469152c141b40593","subject":"Expose version error message generically","status":"MERGED","created":"2019-04-17 05:30:58.000000000","updated":"2019-05-20 13:23:40.000000000","submitted":"2019-05-20 13:23:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653285-1558358620729-a663f06a","meta_rev_id":"adb969ebf269d38e6790a3c8c767ddc0e8d58d26","_number":653285,"virtual_id_number":653285,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-20 13:10:26.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-05-20 13:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-05-20 13:04:25.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 11:44:13.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-05-20 13:04:25.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 17:07:37.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-05-20 13:04:25.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-05-20 13:10:26.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-20 13:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d3e79b07dbdc23591d283d1b25758eea4793b2d","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-17 05:30:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9c05d87d8f059e2b3530916b397b0cdccf10771","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 06:04:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/653285/1/check/openstack-tox-pep8/561eecd/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/85/653285/1/check/openstack-tox-py27/e58f69f/ : SUCCESS in 5m 04s\n- openstack-tox-py35 http://logs.openstack.org/85/653285/1/check/openstack-tox-py35/5891f30/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/85/653285/1/check/openstack-tox-py36/acc7840/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/85/653285/1/check/openstack-tox-py37/6678fef/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/85/653285/1/check/openstack-tox-docs/75a82ff/html/ : SUCCESS in 5m 49s\n- osc-placement-dsvm-functional http://logs.openstack.org/85/653285/1/check/osc-placement-dsvm-functional/98d1057/ : SUCCESS in 19m 41s\n- osc-placement-dsvm-functional-py3 http://logs.openstack.org/85/653285/1/check/osc-placement-dsvm-functional-py3/e23c3e1/ : SUCCESS in 20m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cba5dd2401b18bf87f2e957d50314a4d530f5d70","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-04-17 18:46:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThis is rather lovely. There\u0027s a grammar cleanup within which is probably worth fixing, thus the -1","accounts_in_message":[],"_revision_number":1},{"id":"a08f2f6de9ab773606b9f7008eb963989a7642c4","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-18 04:26:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dab2061d751bd0eaeca1ee151961d399d0514068","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 04:57:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/653285/2/check/openstack-tox-pep8/b15d852/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/85/653285/2/check/openstack-tox-py27/6d47ef6/ : FAILURE in 5m 30s\n- openstack-tox-py35 http://logs.openstack.org/85/653285/2/check/openstack-tox-py35/16315a1/ : FAILURE in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/85/653285/2/check/openstack-tox-py36/0a1dcb6/ : FAILURE in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/85/653285/2/check/openstack-tox-py37/d3e342e/ : FAILURE in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/85/653285/2/check/openstack-tox-docs/a723a54/html/ : SUCCESS in 4m 30s\n- osc-placement-dsvm-functional http://logs.openstack.org/85/653285/2/check/osc-placement-dsvm-functional/60d1fe8/ : SUCCESS in 23m 13s\n- osc-placement-dsvm-functional-py3 http://logs.openstack.org/85/653285/2/check/osc-placement-dsvm-functional-py3/ae00cdf/ : SUCCESS in 22m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e5c2f3675870a328b125b5c5e53c64be263f1389","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-18 07:10:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"237fbf8460274bec81a19442456c1bd1407f8b57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 07:44:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/653285/3/check/openstack-tox-pep8/053710d/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/85/653285/3/check/openstack-tox-py27/687cbbe/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/85/653285/3/check/openstack-tox-py35/8c80108/ : SUCCESS in 6m 17s\n- openstack-tox-py36 http://logs.openstack.org/85/653285/3/check/openstack-tox-py36/5751dd6/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/85/653285/3/check/openstack-tox-py37/91dd0d0/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/85/653285/3/check/openstack-tox-docs/c1f9ae5/html/ : SUCCESS in 6m 33s\n- osc-placement-dsvm-functional http://logs.openstack.org/85/653285/3/check/osc-placement-dsvm-functional/c8b7336/ : SUCCESS in 26m 49s\n- osc-placement-dsvm-functional-py3 http://logs.openstack.org/85/653285/3/check/osc-placement-dsvm-functional-py3/53c134d/ : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":3},{"id":"950bd0a7063383f3cad7bc0e0d861c37f4fa7fd5","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-04-22 05:14:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8ac89c80c481f00aeb969761a0b3edb47bf7ec3","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-10 17:07:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"35ce0ebf60301a680e5f8944d15ebc38ae57e7a3","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-20 11:44:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a6c4f60fdfd0351c972c660bcd7e4f50e935a128","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:56:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/653285/4/check/openstack-tox-pep8/3c0c8ab/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/85/653285/4/check/openstack-tox-py27/154c9e3/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/85/653285/4/check/openstack-tox-py36/8bfe64f/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/85/653285/4/check/openstack-tox-py37/02aa683/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/85/653285/4/check/openstack-tox-docs/709c8eb/html/ : SUCCESS in 3m 53s\n- openstack-tox-functional http://logs.openstack.org/85/653285/4/check/openstack-tox-functional/15c0cb2/ : SUCCESS in 4m 33s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/653285/4/check/openstack-tox-functional-py36/17904cb/ : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":4},{"id":"6f8a67e1b56d28dac3ab2e312df12d8922111db4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-20 13:04:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nlgtm, thanks.","accounts_in_message":[],"_revision_number":4},{"id":"3459133e1bae1d2ff993e1c5e685e4b467aa2c6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 13:04:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bd707d3aff3190a655f5b225de305694fbb4f1e6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-20 13:10:26.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fe71ff518d6c9e4f9fae7ff0e79ef2c78b85609a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 13:23:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"7a770fbd22e5305bf6f3bf26c4a4991c52b239d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 13:23:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/653285/4/gate/openstack-tox-pep8/7e9f011/ : SUCCESS in 6m 02s\n- openstack-tox-py27 http://logs.openstack.org/85/653285/4/gate/openstack-tox-py27/c105f0b/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/85/653285/4/gate/openstack-tox-py36/d9e15ba/ : SUCCESS in 5m 33s\n- openstack-tox-py37 http://logs.openstack.org/85/653285/4/gate/openstack-tox-py37/73e95a4/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/85/653285/4/gate/openstack-tox-docs/5df4efb/html/ : SUCCESS in 3m 53s\n- openstack-tox-functional http://logs.openstack.org/85/653285/4/gate/openstack-tox-functional/b87a9e9/ : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/653285/4/gate/openstack-tox-functional-py36/a7b77f0/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"250a8d18cdc57c7f889aeb1df2a891ce3e48db4e","revisions":{"a2bb0ff2411758fb52f4ddcbd5b4d8a6b04138b7":{"kind":"REWORK","_number":1,"created":"2019-04-17 05:30:58.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/85/653285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/85/653285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/1"}}},"commit":{"parents":[{"commit":"cf0d72191fd608f8203935cd8d17baeb643a0df8","subject":"Improve aggregate version check error messages with min_version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/cf0d72191fd608f8203935cd8d17baeb643a0df8"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-04-17 04:54:51.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-04-17 05:10:50.000000000","tz":0},"subject":"Expose version error message generically","message":"Expose version error message generically\n\nThe change in https://review.openstack.org/#/c/652100/ has improved\nthe error message for the required placement api version::\n\n Operation or argument is not supported with version 1.0\n\nhad become::\n\n Operation or argument is not supported with version 1.0; \\\n  requires at least version 1.1\n\nHowever, this was only applied for aggregate use cases. This patch\nfollows it up to apply the improved error message generically to other\ncommands than aggregate and other version compare function than\n\"greater than or equal to\".\n\nChange-Id: I28226a1dbc003a8b6366e528469152c141b40593\nStory: #2005448\nTask: #30498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a2bb0ff2411758fb52f4ddcbd5b4d8a6b04138b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/a2bb0ff2411758fb52f4ddcbd5b4d8a6b04138b7"}]},"branch":"refs/heads/master"},"e20734d1a7112c619ae5628cf4ea6b8f7bb7c74b":{"kind":"REWORK","_number":2,"created":"2019-04-18 04:26:26.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/85/653285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/85/653285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/2"}}},"commit":{"parents":[{"commit":"cf0d72191fd608f8203935cd8d17baeb643a0df8","subject":"Improve aggregate version check error messages with min_version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/cf0d72191fd608f8203935cd8d17baeb643a0df8"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-04-17 04:54:51.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-04-16 15:17:38.000000000","tz":0},"subject":"Expose version error message generically","message":"Expose version error message generically\n\nThe change in https://review.openstack.org/#/c/652100/ has improved\nthe error message for the required placement api version::\n\n Operation or argument is not supported with version 1.0\n\nhad become::\n\n Operation or argument is not supported with version 1.0; \\\n  requires at least version 1.1\n\nHowever, this was only applied for aggregate use cases. This patch\nfollows it up to apply the improved error message generically to other\ncommands than aggregate and other version compare function than\n\"greater than or equal to\".\n\nChange-Id: I28226a1dbc003a8b6366e528469152c141b40593\nStory: #2005448\nTask: #30498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e20734d1a7112c619ae5628cf4ea6b8f7bb7c74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e20734d1a7112c619ae5628cf4ea6b8f7bb7c74b"}]},"branch":"refs/heads/master"},"58859511e60dd049ebc7d8c8d53e673858167c0b":{"kind":"REWORK","_number":3,"created":"2019-04-18 07:10:16.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/85/653285/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/85/653285/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/3"}}},"commit":{"parents":[{"commit":"cf0d72191fd608f8203935cd8d17baeb643a0df8","subject":"Improve aggregate version check error messages with min_version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/cf0d72191fd608f8203935cd8d17baeb643a0df8"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-04-17 04:54:51.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-04-16 18:00:46.000000000","tz":0},"subject":"Expose version error message generically","message":"Expose version error message generically\n\nThe change in https://review.openstack.org/#/c/652100/ has improved\nthe error message for the required placement api version::\n\n Operation or argument is not supported with version 1.0\n\nhad become::\n\n Operation or argument is not supported with version 1.0; \\\n  requires at least version 1.1\n\nHowever, this was only applied for aggregate use cases. This patch\nfollows it up to apply the improved error message generically to other\ncommands than aggregate and other version compare function than\n\"greater than or equal to\".\n\nChange-Id: I28226a1dbc003a8b6366e528469152c141b40593\nStory: #2005448\nTask: #30498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/58859511e60dd049ebc7d8c8d53e673858167c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/58859511e60dd049ebc7d8c8d53e673858167c0b"}]},"branch":"refs/heads/master"},"250a8d18cdc57c7f889aeb1df2a891ce3e48db4e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-20 11:44:13.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/85/653285/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/85/653285/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/85/653285/4"}}},"commit":{"parents":[{"commit":"da8cd4d68b06399c607776db2a704b4578146996","subject":"Use PlacementFixture in functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/da8cd4d68b06399c607776db2a704b4578146996"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-04-17 04:54:51.000000000","tz":0},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-05-20 11:44:13.000000000","tz":0},"subject":"Expose version error message generically","message":"Expose version error message generically\n\nThe change in https://review.openstack.org/#/c/652100/ has improved\nthe error message for the required placement api version::\n\n Operation or argument is not supported with version 1.0\n\nhad become::\n\n Operation or argument is not supported with version 1.0; \\\n  requires at least version 1.1\n\nHowever, this was only applied for aggregate use cases. This patch\nfollows it up to apply the improved error message generically to other\ncommands than aggregate and other version compare function than\n\"greater than or equal to\".\n\nChange-Id: I28226a1dbc003a8b6366e528469152c141b40593\nStory: #2005448\nTask: #30498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/250a8d18cdc57c7f889aeb1df2a891ce3e48db4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/250a8d18cdc57c7f889aeb1df2a891ce3e48db4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
