)]}'
{"id":"openstack%2Fosc-placement~655057","triplet_id":"openstack%2Fosc-placement~master~I2f690a9910351afd52703e4a0c537fa3b011f778","project":"openstack/osc-placement","branch":"master","topic":"bug/1826823","hashtags":[],"change_id":"I2f690a9910351afd52703e4a0c537fa3b011f778","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:39:46.000000000","updated":"2019-06-13 09:13:06.000000000","submitted":"2019-06-13 09:13:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655057-1560417186218-9e97f1fb","meta_rev_id":"5dd7f644075e8177a49bd7ca72b69f6c3f441414","_number":655057,"virtual_id_number":655057,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-06-13 09:13:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"recommended":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2019-06-12 15:07:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-06-13 08:17:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-13 08:44:23.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-13 08:44:23.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 16:01:47.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-06-04 10:09:16.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-06 11:15:21.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-06-11 07:14:03.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-06-12 15:07:36.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-06-13 08:17:14.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-06-13 08:44:23.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-13 09:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0273d8eaf9d62b79c4ca5f0913c2d822dad8554","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:39:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2b4ec535ca3f140bd3fa4be6a0f2abbbe03c74d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:04:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/57/655057/1/check/test-release-openstack/8246fa5/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/57/655057/1/check/openstack-tox-pep8/95d9ca1/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/57/655057/1/check/openstack-tox-py27/0c0f323/ : SUCCESS in 4m 35s\n- openstack-tox-py35 http://logs.openstack.org/57/655057/1/check/openstack-tox-py35/1c429a0/ : SUCCESS in 6m 53s\n- openstack-tox-py36 http://logs.openstack.org/57/655057/1/check/openstack-tox-py36/046bc6a/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/57/655057/1/check/openstack-tox-py37/c8eda31/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/57/655057/1/check/openstack-tox-docs/26d2fa7/html/ : SUCCESS in 4m 13s\n- build-openstack-releasenotes http://logs.openstack.org/57/655057/1/check/build-openstack-releasenotes/9792c99/html/ : SUCCESS in 3m 57s\n- osc-placement-dsvm-functional http://logs.openstack.org/57/655057/1/check/osc-placement-dsvm-functional/afb396b/ : FAILURE in 19m 41s\n- osc-placement-dsvm-functional-py3 http://logs.openstack.org/57/655057/1/check/osc-placement-dsvm-functional-py3/bf6103a/ : FAILURE in 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"80e90a8c5c247dcb543f389fd255c3e14ab12351","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-04-24 01:58:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"980a270b0ff93f5a5196422bdffb824e1280ee64","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-04-26 16:01:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e9066098278c7575f68bd36727008456fe64a76","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-29 05:11:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cbe4f731f6a696883144e124b3e87787692b475b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 05:35:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/57/655057/2/check/test-release-openstack/be69424/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/57/655057/2/check/openstack-tox-pep8/04d70e5/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/57/655057/2/check/openstack-tox-py27/2e98d02/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/57/655057/2/check/openstack-tox-py35/b54c9ea/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/57/655057/2/check/openstack-tox-py36/4b89611/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/57/655057/2/check/openstack-tox-py37/12c2767/ : SUCCESS in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/57/655057/2/check/openstack-tox-docs/3b813ca/html/ : SUCCESS in 3m 50s\n- build-openstack-releasenotes http://logs.openstack.org/57/655057/2/check/build-openstack-releasenotes/570f351/html/ : SUCCESS in 3m 36s\n- osc-placement-dsvm-functional http://logs.openstack.org/57/655057/2/check/osc-placement-dsvm-functional/d9ae956/ : SUCCESS in 19m 03s\n- osc-placement-dsvm-functional-py3 http://logs.openstack.org/57/655057/2/check/osc-placement-dsvm-functional-py3/e76a73b/ : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":2},{"id":"88bbb0f7104408182ac3ceaed997b91e3206555e","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-04-30 13:03:17.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"e5ae06c7045727617d956e2616dff1619a4890da","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-03 03:23:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f8c279760961000451a9479cb5a84664841780ca","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-10 16:56:59.000000000","message":"Patch Set 2: Code-Review-1\n\nThis duplicates https://review.opendev.org/#/c/658270/ and has the same issues mentioned there. Please fix one of them and we\u0027ll get it merged.","accounts_in_message":[],"_revision_number":2},{"id":"3fa016dcb8f6ef3fb889dbe124d0b841fde23c39","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-11 04:09:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0bc930b30efe1972b286dba2b8f63a4dbab2c76a","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-11 04:13:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"28763ec3b06efb563ee75e9b1a679d16609c99a3","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-11 04:15:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"03f66ffbf5621c6d592581b8d0dea930ffdab7c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 04:52:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/57/655057/5/check/test-release-openstack/1b4c511/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/57/655057/5/check/openstack-tox-pep8/0a40e14/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/57/655057/5/check/openstack-tox-py27/7954ca4/ : SUCCESS in 4m 21s\n- openstack-tox-py35 http://logs.openstack.org/57/655057/5/check/openstack-tox-py35/508d2c2/ : SUCCESS in 6m 46s\n- openstack-tox-py36 http://logs.openstack.org/57/655057/5/check/openstack-tox-py36/15c7c02/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/57/655057/5/check/openstack-tox-py37/8c406f0/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/57/655057/5/check/openstack-tox-docs/00631fb/ : POST_FAILURE in 3m 36s\n- build-openstack-releasenotes http://logs.openstack.org/57/655057/5/check/build-openstack-releasenotes/404b47e/html/ : SUCCESS in 3m 55s\n- openstack-tox-functional http://logs.openstack.org/57/655057/5/check/openstack-tox-functional/4f5599d/ : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/655057/5/check/openstack-tox-functional-py36/603f6e9/ : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":5},{"id":"7b63089d55ca5e3cb02c7cf7df136cd54bb660e2","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-20 11:43:11.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6d4278d9584f380c76793e46bbfd4f0831d18d6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 12:04:27.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/57/655057/5/check/test-release-openstack/5e79bab/ : SUCCESS in 2m 57s\n- openstack-tox-pep8 http://logs.openstack.org/57/655057/5/check/openstack-tox-pep8/51a0b26/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/57/655057/5/check/openstack-tox-py27/30fdfc3/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/57/655057/5/check/openstack-tox-py36/d606687/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/57/655057/5/check/openstack-tox-py37/bdd0a60/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/57/655057/5/check/openstack-tox-docs/0297af7/ : POST_FAILURE in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/57/655057/5/check/build-openstack-releasenotes/4d8da4f/html/ : SUCCESS in 3m 48s\n- openstack-tox-functional http://logs.openstack.org/57/655057/5/check/openstack-tox-functional/d729be5/ : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/655057/5/check/openstack-tox-functional-py36/8403fe2/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":5},{"id":"ca5b04cfcd91e78dcf8f06d549cee802e200ac55","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-04 10:09:16.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"609e86cd68232a7a39b64a2e8a5afc169c3b949e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-06 11:14:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"495140d55e4b8ec2c293cd03fd8aee3665e0d475","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-06 11:15:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"045ae2e08686a080cb6b8589ec26aece1bd2bf5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 11:47:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/57/655057/5/check/test-release-openstack/fe21faa/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/57/655057/5/check/openstack-tox-pep8/13d0b3a/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/57/655057/5/check/openstack-tox-py27/87ec16a/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/57/655057/5/check/openstack-tox-py36/c127b3b/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/57/655057/5/check/openstack-tox-py37/3bdda00/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/57/655057/5/check/openstack-tox-docs/534efc8/ : POST_FAILURE in 3m 51s\n- build-openstack-releasenotes http://logs.openstack.org/57/655057/5/check/build-openstack-releasenotes/f79c18a/html/ : SUCCESS in 3m 40s\n- openstack-tox-functional http://logs.openstack.org/57/655057/5/check/openstack-tox-functional/c7f3ee5/ : SUCCESS in 4m 36s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/655057/5/check/openstack-tox-functional-py36/924ca9a/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":5},{"id":"d5600e8f13c8062cd45215e3e59ccf39586e2ae1","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-06-11 07:14:03.000000000","message":"Patch Set 5: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ba966c2435d9c85724cc5c1a2f1644245cb5975b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 07:27:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/57/655057/5/check/test-release-openstack/613f00e/ : SUCCESS in 3m 28s\n- openstack-tox-pep8 http://logs.openstack.org/57/655057/5/check/openstack-tox-pep8/14fe387/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/57/655057/5/check/openstack-tox-py27/a21f640/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/57/655057/5/check/openstack-tox-py36/eb8cefa/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/57/655057/5/check/openstack-tox-py37/45d5325/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/57/655057/5/check/openstack-tox-docs/a408305/ : POST_FAILURE in 3m 46s\n- build-openstack-releasenotes http://logs.openstack.org/57/655057/5/check/build-openstack-releasenotes/717c431/html/ : SUCCESS in 4m 23s\n- openstack-tox-functional http://logs.openstack.org/57/655057/5/check/openstack-tox-functional/d21f88f/ : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/655057/5/check/openstack-tox-functional-py36/448bcd9/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":5},{"id":"fac3f20880cdf8e23b10462a4527b2de07294f62","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-11 08:51:12.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f950a4328b5de8d40f41b57b8f2a2c7e4d0b524f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-11 09:11:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"95306248d2fef8b432f054f413503a3aad68557d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 09:34:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/57/655057/6/check/test-release-openstack/ced4e12/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/57/655057/6/check/openstack-tox-pep8/f3f6890/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/57/655057/6/check/openstack-tox-py27/75dfaed/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/57/655057/6/check/openstack-tox-py36/76d56d8/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/57/655057/6/check/openstack-tox-py37/3e5b5ce/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/57/655057/6/check/openstack-tox-docs/17f6e61/html/ : SUCCESS in 4m 20s\n- build-openstack-releasenotes http://logs.openstack.org/57/655057/6/check/build-openstack-releasenotes/212672d/html/ : SUCCESS in 3m 58s\n- openstack-tox-functional http://logs.openstack.org/57/655057/6/check/openstack-tox-functional/ac2a7cd/ : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/655057/6/check/openstack-tox-functional-py36/a32c9a9/ : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":6},{"id":"ff12a3b19216648987aedb3b3360313fe872f8ef","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-12 15:07:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"916f786c5d027024d5b57e16eaf85f6bf00954cf","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-13 08:17:14.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"246ceedbae840f1fdc869527db4cc45413cd34cf","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-13 08:44:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)\n\nrestoring gibi\u0027s approval","accounts_in_message":[],"_revision_number":6},{"id":"0737035f1cd150be023ecf86346d4798d2b42b51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 08:44:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"127e9bba2cc1e50a67c4a0f8b27ae88df06715b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:13:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"4c527e22d870be71267b0d20216e2823fac071dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:13:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/57/655057/6/gate/test-release-openstack/1156c9f/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/57/655057/6/gate/openstack-tox-pep8/bce4023/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/57/655057/6/gate/openstack-tox-py27/d27563b/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/57/655057/6/gate/openstack-tox-py36/b329cb5/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/57/655057/6/gate/openstack-tox-py37/21862e8/ : SUCCESS in 6m 08s\n- openstack-tox-docs http://logs.openstack.org/57/655057/6/gate/openstack-tox-docs/5611591/html/ : SUCCESS in 4m 16s\n- build-openstack-releasenotes http://logs.openstack.org/57/655057/6/gate/build-openstack-releasenotes/5f8561f/html/ : SUCCESS in 4m 09s\n- openstack-tox-functional http://logs.openstack.org/57/655057/6/gate/openstack-tox-functional/24e4d85/ : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 http://logs.openstack.org/57/655057/6/gate/openstack-tox-functional-py36/b123858/ : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"78b1f3ea8239ad06b9bf9c9abb78f00707a805a7","revisions":{"d2ace2a5f21341a9aa5bad4c52a98d56ea7c8e4c":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:39:46.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/57/655057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/57/655057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/1"}}},"commit":{"parents":[{"commit":"f79dbf7277c323adc0f04b21ba8f290b22841018","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f79dbf7277c323adc0f04b21ba8f290b22841018"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:24.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:43.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I2f690a9910351afd52703e4a0c537fa3b011f778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d2ace2a5f21341a9aa5bad4c52a98d56ea7c8e4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d2ace2a5f21341a9aa5bad4c52a98d56ea7c8e4c"}]},"branch":"refs/heads/master"},"29e64e9abac73ef9d6ad227b34f040c36a3fc8e5":{"kind":"REWORK","_number":2,"created":"2019-04-29 05:11:02.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/57/655057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/57/655057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/2"}}},"commit":{"parents":[{"commit":"f79dbf7277c323adc0f04b21ba8f290b22841018","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f79dbf7277c323adc0f04b21ba8f290b22841018"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:24.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-29 05:10:43.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I2f690a9910351afd52703e4a0c537fa3b011f778\nCloses-Bug: #1826823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/29e64e9abac73ef9d6ad227b34f040c36a3fc8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/29e64e9abac73ef9d6ad227b34f040c36a3fc8e5"}]},"branch":"refs/heads/master"},"5e26088fe141c211d9fb5e54de132b44427b375d":{"kind":"REWORK","_number":3,"created":"2019-05-11 04:09:04.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/57/655057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/57/655057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/3"}}},"commit":{"parents":[{"commit":"f79dbf7277c323adc0f04b21ba8f290b22841018","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f79dbf7277c323adc0f04b21ba8f290b22841018"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:24.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-11 04:09:01.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I2f690a9910351afd52703e4a0c537fa3b011f778\nCloses-Bug: #1826823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5e26088fe141c211d9fb5e54de132b44427b375d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5e26088fe141c211d9fb5e54de132b44427b375d"}]},"branch":"refs/heads/master"},"21612a5cb347a7bb4e52b6b821f93c91be93813f":{"kind":"REWORK","_number":4,"created":"2019-05-11 04:13:33.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/57/655057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/57/655057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/4"}}},"commit":{"parents":[{"commit":"f79dbf7277c323adc0f04b21ba8f290b22841018","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f79dbf7277c323adc0f04b21ba8f290b22841018"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:24.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-11 04:13:19.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I2f690a9910351afd52703e4a0c537fa3b011f778\nCloses-Bug: #1826823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/21612a5cb347a7bb4e52b6b821f93c91be93813f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/21612a5cb347a7bb4e52b6b821f93c91be93813f"}]},"branch":"refs/heads/master"},"495fa5a5b4dccaabeb33743299f22da45e72fa29":{"kind":"REWORK","_number":5,"created":"2019-05-11 04:15:36.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/57/655057/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/57/655057/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/5"}}},"commit":{"parents":[{"commit":"f79dbf7277c323adc0f04b21ba8f290b22841018","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f79dbf7277c323adc0f04b21ba8f290b22841018"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:24.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-11 04:15:28.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I2f690a9910351afd52703e4a0c537fa3b011f778\nCloses-Bug: #1826823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/495fa5a5b4dccaabeb33743299f22da45e72fa29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/495fa5a5b4dccaabeb33743299f22da45e72fa29"}]},"branch":"refs/heads/master"},"78b1f3ea8239ad06b9bf9c9abb78f00707a805a7":{"kind":"REWORK","_number":6,"created":"2019-06-11 09:11:54.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/57/655057/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/57/655057/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/57/655057/6"}}},"commit":{"parents":[{"commit":"e52fa969f9ee7ff2e58ce86c2a6261a80b819093","subject":"Add support for 1.22 microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:24.000000000","tz":480},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-11 09:11:20.000000000","tz":60},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nAlso fix some remaining inconsistencies in tox.ini\n\nChange-Id: I2f690a9910351afd52703e4a0c537fa3b011f778\nCloses-Bug: #1826823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/78b1f3ea8239ad06b9bf9c9abb78f00707a805a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/78b1f3ea8239ad06b9bf9c9abb78f00707a805a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
