)]}'
{"id":"openstack%2Fosc-placement~666542","triplet_id":"openstack%2Fosc-placement~master~I0f3068c411c504380918f27f994734528fe55c77","project":"openstack/osc-placement","branch":"master","topic":"catch-up-microversion","hashtags":[],"change_id":"I0f3068c411c504380918f27f994734528fe55c77","subject":"Support multiple member_of query parameter","status":"MERGED","created":"2019-06-20 11:27:41.000000000","updated":"2021-03-19 15:28:06.000000000","submitted":"2021-03-19 15:22:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"666542-catch-up-microversion","meta_rev_id":"ae5e8a68a4ffc3ce2e4bbc9c2b2a575144dd7ac5","_number":666542,"virtual_id_number":666542,"owner":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-19 15:22:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2021-03-19 15:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2020-09-11 06:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2021-03-19 15:03:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 11:33:10.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-20 14:10:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-07-03 12:03:42.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-07-03 14:21:04.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-07-03 15:59:37.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2020-01-14 06:12:55.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2020-01-14 10:37:08.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-09-11 07:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 15:03:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"76fdbae1fae0c65312e73926dfb1b22aa889367e","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-06-20 11:27:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fddeaa96ddc6b9592ce678cec68b708d33000213","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-06-20 11:30:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1199baf9f5472ff71354000d45031ec4f93925b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 11:47:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/42/666542/2/check/openstack-tox-pep8/f37d208/ : FAILURE in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/42/666542/2/check/openstack-tox-py27/72c32f3/ : POST_FAILURE in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/42/666542/2/check/openstack-tox-py36/0ab8bb7/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/42/666542/2/check/openstack-tox-py37/a1e58c1/ : SUCCESS in 10m 14s\n- openstack-tox-docs http://logs.openstack.org/42/666542/2/check/openstack-tox-docs/7269f38/html/ : SUCCESS in 9m 33s\n- build-openstack-releasenotes http://logs.openstack.org/42/666542/2/check/build-openstack-releasenotes/39e38a1/html/ : SUCCESS in 4m 10s\n- openstack-tox-functional http://logs.openstack.org/42/666542/2/check/openstack-tox-functional/08c85fa/ : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/666542/2/check/openstack-tox-functional-py36/b47a3ed/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"42b01e8492932875ec50ea4be652280af922836b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-06-20 14:10:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b1ad65d9b8abdc79d9c65b559a23ac3747513bb4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-20 15:17:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nI think the meaning of the 1.24 microversion has been misunderstood, so this implementation is not quite right. I\u0027ve tried to provide some links in comments within to additional sources of information which may help explain how it is supposed to work.","accounts_in_message":[],"_revision_number":2},{"id":"bd487c3183baed31a6873162de42a73d4d018511","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-06-21 13:17:08.000000000","message":"Patch Set 2:\n\nThank you for every comment. I can assure that I understand the idea of usage multiple member_of query parameters, it\u0027s just the descriptions that need to be improved, and the code of course, but for now let\u0027s please discuss if you like the idea of asking user to use \u0027in:\u0027 keyword when they would like to use it the way that was introduced in 1.24.\n\nThe code that I have written so far, works only when multiple --aggregate-uuid options are used, so:\n\nstack@baclbpaw-devstack:~/osc-placement$ openstack --os-placement-api-version 1.24 resource provider list --aggregate-uuid in:35819442-6414-46d5-a0d8-d81c6c995938\n\nURL: /resource_providers?member_of\u003din%3A35819442-6414-46d5-a0d8-d81c6c995938\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| uuid                                 | name | generation | root_provider_uuid                   | parent_provider_uuid |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| 434e9771-d2bb-4b02-a862-268f88923d98 | rp1  |          5 | 434e9771-d2bb-4b02-a862-268f88923d98 | None                 |\n| 6a2a1d33-c5c6-4563-9159-6071ce109660 | rp2  |          2 | 6a2a1d33-c5c6-4563-9159-6071ce109660 | None                 |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n\nstack@baclbpaw-devstack:~/osc-placement$ openstack --os-placement-api-version 1.24 resource provider list --aggregate-uuid in:7bc7318f-ffb3-4e7e-8e20-9965d2292849\n\nURL: /resource_providers?member_of\u003din%3A7bc7318f-ffb3-4e7e-8e20-9965d2292849\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| uuid                                 | name | generation | root_provider_uuid                   | parent_provider_uuid |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| 6a2a1d33-c5c6-4563-9159-6071ce109660 | rp2  |          2 | 6a2a1d33-c5c6-4563-9159-6071ce109660 | None                 |\n| e094a039-372b-44f4-b3e5-2e6a1858e4f0 | rp3  |          0 | e094a039-372b-44f4-b3e5-2e6a1858e4f0 | None                 |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n\nand when used multiple --aggregate-uuid options:\n\nstack@baclbpaw-devstack:~/osc-placement$ openstack --os-placement-api-version 1.24 resource provider list --aggregate-uuid in:35819442-6414-46d5-a0d8-d81c6c995938 --aggregate-uuid in:7bc7318f-ffb3-4e7e-8e20-9965d2292849\n\nURL: /resource_providers?member_of\u003din%3A35819442-6414-46d5-a0d8-d81c6c995938\u0026member_of\u003din%3A7bc7318f-ffb3-4e7e-8e20-9965d2292849\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| uuid                                 | name | generation | root_provider_uuid                   | parent_provider_uuid |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| 6a2a1d33-c5c6-4563-9159-6071ce109660 | rp2  |          2 | 6a2a1d33-c5c6-4563-9159-6071ce109660 | None                 |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n\nstack@baclbpaw-devstack:~/osc-placement$ openstack --os-placement-api-version 1.24 resource provider list --aggregate-uuid in:35819442-6414-46d5-a0d8-d81c6c995938,7bc7318f-ffb3-4e7e-8e20-9965d2292849 --aggregate-uuid in:7bc7318f-ffb3-4e7e-8e20-9965d2292849\n\nURL: /resource_providers?member_of\u003din%3A35819442-6414-46d5-a0d8-d81c6c995938%2C7bc7318f-ffb3-4e7e-8e20-9965d2292849\u0026member_of\u003din%3A7bc7318f-ffb3-4e7e-8e20-9965d2292849\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| uuid                                 | name | generation | root_provider_uuid                   | parent_provider_uuid |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| 6a2a1d33-c5c6-4563-9159-6071ce109660 | rp2  |          2 | 6a2a1d33-c5c6-4563-9159-6071ce109660 | None                 |\n| e094a039-372b-44f4-b3e5-2e6a1858e4f0 | rp3  |          0 | e094a039-372b-44f4-b3e5-2e6a1858e4f0 | None                 |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n\n\nand backward compatibility (no \u0027in:\u0027 keyword results with single member_of query parameter):\n\nstack@baclbpaw-devstack:~/osc-placement$ openstack --os-placement-api-version 1.24 resource provider list --aggregate-uuid 35819442-6414-46d5-a0d8-d81c6c995938 --aggregate-uuid 7bc7318f-ffb3-4e7e-8e20-9965d2292849\n\nURL: /resource_providers?member_of\u003din%3A35819442-6414-46d5-a0d8-d81c6c995938%2C7bc7318f-ffb3-4e7e-8e20-9965d2292849\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| uuid                                 | name | generation | root_provider_uuid                   | parent_provider_uuid |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n| 434e9771-d2bb-4b02-a862-268f88923d98 | rp1  |          5 | 434e9771-d2bb-4b02-a862-268f88923d98 | None                 |\n| 6a2a1d33-c5c6-4563-9159-6071ce109660 | rp2  |          2 | 6a2a1d33-c5c6-4563-9159-6071ce109660 | None                 |\n| e094a039-372b-44f4-b3e5-2e6a1858e4f0 | rp3  |          0 | e094a039-372b-44f4-b3e5-2e6a1858e4f0 | None                 |\n+--------------------------------------+------+------------+--------------------------------------+----------------------+\n\nWith the code I have writter, we cannot use \u0027in:\u0027 multiple times at once, so --aggregate-uuid in:35819442-6414-46d5-a0d8-d81c6c995938,in:7bc7318f-ffb3-4e7e-8e20-9965d2292849 won\u0027t work, but it can be easily improved.\n\nFor now, I just want to know your opinion on that, so I know if I should proceed with this idea.","accounts_in_message":[],"_revision_number":2},{"id":"0316639ababf476e022e3c4b081356ce3d06b619","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-06-25 07:42:08.000000000","message":"Patch Set 2:\n\n@Pawel: I think the best is the release notes sent by Chris: https://review.opendev.org/#/c/561315/16/releasenotes/notes/multi-member-of-4f9518a96652c0c6.yaml, and do the testing based on that, hope that placement folks agree with me :-)","accounts_in_message":[],"_revision_number":2},{"id":"4a55740bfdbf95740b8adddbdbe64ef6167520e5","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-06-26 08:03:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"17e8a18d45b2896303dcecc267ce12e12de04341","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-06-26 08:09:22.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1a9ae48033999790f03d726e53a024f37f0e1a16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:20:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/42/666542/3/check/openstack-tox-pep8/33cceaa/ : FAILURE in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/42/666542/3/check/openstack-tox-py27/3cc8825/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/42/666542/3/check/openstack-tox-py36/dfcb1d5/ : SUCCESS in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/42/666542/3/check/openstack-tox-py37/88167b2/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/42/666542/3/check/openstack-tox-docs/1e464ef/html/ : SUCCESS in 3m 00s\n- build-openstack-releasenotes http://logs.openstack.org/42/666542/3/check/build-openstack-releasenotes/28667cf/ : FAILURE in 2m 46s\n- openstack-tox-functional http://logs.openstack.org/42/666542/3/check/openstack-tox-functional/703e49c/ : SUCCESS in 3m 33s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/666542/3/check/openstack-tox-functional-py36/dc5b279/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"c68bef5f5c6452db6ab25bd3c6fff601b50e3ffd","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-06-26 10:09:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1f4370cca5eb033a49f461f3808295a47ce49c99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 10:39:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/42/666542/4/check/openstack-tox-pep8/bb77524/ : FAILURE in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/42/666542/4/check/openstack-tox-py27/122f74c/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/42/666542/4/check/openstack-tox-py36/18f5108/ : SUCCESS in 3m 06s\n- openstack-tox-py37 http://logs.openstack.org/42/666542/4/check/openstack-tox-py37/21e1455/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/42/666542/4/check/openstack-tox-docs/e0a7105/ : POST_FAILURE in 15m 06s\n- build-openstack-releasenotes http://logs.openstack.org/42/666542/4/check/build-openstack-releasenotes/7770b64/html/ : SUCCESS in 2m 39s\n- openstack-tox-functional http://logs.openstack.org/42/666542/4/check/openstack-tox-functional/382162d/ : FAILURE in 15m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/666542/4/check/openstack-tox-functional-py36/67b5141/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"cc7bc40f95604317a6fdbfe167da28050801b23d","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-06-26 11:50:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0d7930de199da991a30f47d84d210eaaa90c120c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 12:02:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/42/666542/5/check/openstack-tox-pep8/dd71b22/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/42/666542/5/check/openstack-tox-py27/fcf0d2f/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/42/666542/5/check/openstack-tox-py36/3d741f1/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/42/666542/5/check/openstack-tox-py37/14755fb/ : SUCCESS in 3m 24s\n- openstack-tox-docs http://logs.openstack.org/42/666542/5/check/openstack-tox-docs/6d17c7f/html/ : SUCCESS in 2m 29s\n- build-openstack-releasenotes http://logs.openstack.org/42/666542/5/check/build-openstack-releasenotes/a49f2e1/html/ : SUCCESS in 2m 59s\n- openstack-tox-functional http://logs.openstack.org/42/666542/5/check/openstack-tox-functional/57d4e8d/ : SUCCESS in 3m 49s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/666542/5/check/openstack-tox-functional-py36/12d7d75/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":5},{"id":"0d453fb7aa1db36bd5d46bbd233005a167bf0244","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-01 13:29:56.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nI think there may still be some confusion on what microversion 1.24 really means, there are too many instances of \"in:\" in some of the examples.\n\nI think one way to address that would be to add a unit or otherwise test that confirms how input parameters on the command line are translated into multiple member_of query parameters.\n\nPerhaps I\u0027m misunderstanding, though, so I\u0027ll refrain from voting at this point until I\u0027m able to look into this more closely.","accounts_in_message":[],"_revision_number":5},{"id":"49979543bc0d1db9e68a8cf3a910c968ec806354","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-07-02 05:50:13.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nI understand that it might seem confusing, but in my opinion it is actually pretty simple.\n\nBut, if anyone has any better concept, how to introduce the 1.24 option, I am open to your suggestions.","accounts_in_message":[],"_revision_number":5},{"id":"5eb04b704e9d1e097fc9142ac6bd9b69af20522f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-03 12:03:42.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cc9d34e42be89e8442e145a3a42d74d4167590db","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-07-03 12:46:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"629452210bde53d6a719c6cf9905d4a977497660","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-07-03 13:26:11.000000000","message":"Patch Set 5:\n\nThe basic problem is that that the CLI modeled the member_of query param with aggregate-uuid CLI param in a non future proof way. \n\nBefore 1.24 the --aggregate-uuid \u003cuuid1\u003e --aggregate-uuid \u003cuuid2\u003e was translated to member_of\u003din:uuid1,uuid2 on the REST API.\n\nAfter 1.24 the REST API uses _repeated_ member_of params to express AND relationship BUT repeating --aggregate-uuid already used up on the CLI to express OR relationship. So the CLI is not easy to extend to the direction that the REST API was extended.\n\nOne option, Pawel proposes, is to introduce the \u0027in:\u0027 prefix to the CLI param. This means that what was worked without \u0027in:\u0027 in the CLI before 1.24 will only work with \u0027in:\u0027 prefix with 1.24. This step synchronizes the way how the REST API and the CLI express OR (with \u0027in:\u0027) and AND (with repeated key) relationship. However it could be confusing for the CLI user how used repeated --aggregate-uuid key before 1.24.  :/\n\nAnother option, I can imagine, is to introduce a totally new CLI param key to get the REST API and the CLI in sync. E.g. we can introduce --member-of key in the CLI where --member-of in:\u003cuuid1\u003e,\u003cuuid2\u003e is translated to member_of\u003d\u0027in:uuid1,uuid2\u0027 on the REST API. And --member-of uuid1 --member-of\u003duuid2 translated to member_of\u003duuid1\u0026member_of\u003duuid2 on the REST API. --member-of CLI param would only be supported with 1.24 or higher microversion of course.\nThis option bad in a sense that there will be now two ways to express aggregate membership, the old and limited --aggregate-uuid CLI param, and the new more flexible --member-of CLI param. We could decide to deprecate and later remove the --aggregate-uuid key to clean up the confusion.\n\n@Chris: do you have a preference?","accounts_in_message":[],"_revision_number":5},{"id":"6fda775c8b09e8010aba624b8540fe2f60eb5c47","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-03 13:31:12.000000000","message":"Patch Set 5:\n\n\u003e The basic problem is that that the CLI modeled the member_of query\n \u003e param with aggregate-uuid CLI param in a non future proof way.\n\nAh, okay, I get it now. Thanks for expressing it in a different way, very helpful.\n\n \u003e Another option, I can imagine, is to introduce a totally new CLI\n \u003e param key to get the REST API and the CLI in sync. E.g. we can\n \u003e introduce --member-of key in the CLI where --member-of\n \u003e in:\u003cuuid1\u003e,\u003cuuid2\u003e is translated to member_of\u003d\u0027in:uuid1,uuid2\u0027 on\n \u003e the REST API. And --member-of uuid1 --member-of\u003duuid2 translated to\n \u003e member_of\u003duuid1\u0026member_of\u003duuid2 on the REST API. --member-of CLI\n \u003e param would only be supported with 1.24 or higher microversion of\n \u003e course.\n \u003e This option bad in a sense that there will be now two ways to\n \u003e express aggregate membership, the old and limited --aggregate-uuid\n \u003e CLI param, and the new more flexible --member-of CLI param. We\n \u003e could decide to deprecate and later remove the --aggregate-uuid key\n \u003e to clean up the confusion.\n\nI prefer this option, but that\u0027s because I prefer there be close parity between the arguments used in URLs and arguments used on the command line. That tends to not be aligned with the ethos of OSC, though, so I\u0027m going to add Dean to the review to get his input on any alternate ideas.\n\nTo be clear, since we talk about member_of if the API documents and the \"how to use placement\" documents, it, to me, is the better argument name.","accounts_in_message":[],"_revision_number":5},{"id":"220edf42eb7db4a38098d11e4e8839b5d04d7072","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-07-03 14:21:04.000000000","message":"Patch Set 5:\n\nCLI arguments like \"--member-of in:\u003cuuid1\u003e:\u003cuuid\u003e\" give me hives.\n\nPassing complex need-to-be-parsed data structures is hard to deal with on the CLI (although makes sense in the REST API itself) - and invariably gets really brittle (see anything related to block_device_mapping).\n\nI\u0027d rather see multiple options - like --member-of \u003cuuid1\u003e --member-of       to make the \u0026\u0026 case like mentioned, but then maybe like --member-in \u003cuuid\u003e --member-in \u003cuuid2\u003e to construct member_of\u003d\"in:uuid1:uuid2\" ... but that\u0027s just off the top of my head, and I\u0027m not sure how weird that would be to actually use. Or, heck - a --member-in \u003cuuid1\u003e,\u003cuuid2\u003e would bother me less than the in: prefix.","accounts_in_message":[],"_revision_number":5},{"id":"05b7781d0b33a5ce5a7f9d7dc3c04fad7c09ffc4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-03 14:38:09.000000000","message":"Patch Set 5:\n\nThere\u0027s some useful discussion in IRC starting near: http://eavesdrop.openstack.org/irclogs/%23openstack-sdks/%23openstack-sdks.2019-07-03.log.html#t2019-07-03T13:41:59\n\nSome of it replicates what has been written here.\n\nI\u0027m leaning towards \"people who want advanced filtering functionality should perhaps be able to pass a raw filter query, exactly as it would be formed in the URL\" but as I\u0027ve said before I don\u0027t think I should decide: I\u0027m not a representative user.\n\nThanks for sticking with this Pawel. Sorry it is dragging.","accounts_in_message":[],"_revision_number":5},{"id":"28df0055be78b9eea87dc3b6957ae877c0352e0b","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2019-07-03 15:59:37.000000000","message":"Patch Set 5:\n\nChris is right in that I consider the UX of the CLI way ahead of modeling the REST API directly.  The CLI serves the user not the API.\n\nThe implied AND/OR operators are the major point of confusion that I see here, especially since they are apparently inconsistent between the existing CLI and new REST API functions.\n\nOne example OSC does have is the --property filter in image list, it is a simple AND when multiple options are present.  There is no OR operator.  In other places where multiple occurrences of an option are used they are also AND-ed, e.g. --property in create \u0026 set commands.\n\nIf compound logical expressions are really required for list command filtering at the CLI consider defining a small DSL that looks to the user like a compound logical expression and passing that in a single --filter option:\n\n  --filter \u0027arg1 and (arg2 or arg3)\u0027","accounts_in_message":[],"_revision_number":5},{"id":"8d3d3bbc0ba8936d8243dd4f5af5889876a4d747","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-07-18 09:50:43.000000000","message":"Patch Set 5:\n\nSo OSC uses repeated params to express AND relationship. Therefore we have an inconsistency in osc-placement as repeated --aggregate-uuid param today is used to express an OR relationship.\n\n\nCan we agree that the way forward is to replace (with deprecation) the --aggregate-uuid param with something better?\n\nI can agree with Monty suggestion of the following\n\n  --member-of \u003cuuid1\u003e --member-of \u003cuuid2\u003e --member-in \u003cuuid3\u003e, \u003cuuid4\u003e\n\nto mean that each returned candidate RP needs to be member of the aggregate: uuid1 AND uuid2 AND (uuid3 OR uuid4)\n\nOr we can design the following based on Dean\u0027s suggestion\n\n  --member-of-filter \u0027\u003cuuid1\u003e and \u003cuuid2\u003e and (\u003cuuid3\u003e or \u003cuuid4\u003e)\u0027\n\nto mean the same as above. However we might need to limit this filter expression because:\n\n  --member-of-filter \u0027(\u003cuuid1\u003e and \u003cuuid2\u003e) or (\u003cuuid3\u003e and \u003cuuid4\u003e)\u0027\n\ncannot be expressed with the current REST API directly but first needs to be translated to:\n\n  (\u003cuuid1\u003e or \u003cuuid3\u003e) and (\u003cuuid2\u003e or \u003cuuid3\u003e) and (\u003cuuid1\u003e or \u003cuuid4\u003e) and (\u003cuuid2\u003e and \u003cuuid4\u003e)\n\nI personally would go with Monty\u0027s suggestion.","accounts_in_message":[],"_revision_number":5},{"id":"98d747d23fe654f7a05c802ba208012f60022898","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-07-19 09:06:09.000000000","message":"Patch Set 5:\n\nThank you Gibi for the clearer explanation and thank you all for the discussion. I have read the IRC discussion as well.\n\nI would also go with Monty\u0027s idea as it seems to be easier to understand for final user that is \u0027not very experienced\u0027 and the \u0027advanced user\u0027 who is able to create some advanced filter expressions, should be also fine with it.","accounts_in_message":[],"_revision_number":5},{"id":"a08e624c1901c386fb94bf033da62167dcdf6cb1","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-22 09:32:29.000000000","message":"Patch Set 5:\n\n\u003e Can we agree that the way forward is to replace (with deprecation)\n \u003e the --aggregate-uuid param with something better?\n \u003e \n \u003e I can agree with Monty suggestion of the following\n \u003e \n \u003e --member-of \u003cuuid1\u003e --member-of \u003cuuid2\u003e --member-in \u003cuuid3\u003e,\n \u003e \u003cuuid4\u003e\n\nThis seems like a good way forward.","accounts_in_message":[],"_revision_number":5},{"id":"5453312126b4d24aba0bdf370be3065039cbb2f7","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-07-30 13:32:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b074e049d76f75c571d26a599b0d8d850405033c","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-07-30 13:42:28.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nDuring work on this feature, I realized that single --member-of option would be enough and less confusing and it would work exactly as in API, so:\n\n--member-of agg1\n\nwould return resource provider that is associated with agg1,\n\n--member-of agg1,agg2\n\nwould return resource provider that are associated with agg1 OR agg2,\n\n--member-of agg1 --member-of agg2\n\nwould return resource provider that are associated with agg1 AND agg2\n\n--member-of agg1,agg2 --member-of agg3\n\nwould return resource provider that are associated with (agg1 OR agg2) AND with agg3\n\nWhat is your opinion on this?","accounts_in_message":[],"_revision_number":6},{"id":"0177a894d39c9054a864af8681923da90aab9eb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 13:51:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/42/666542/6/check/openstack-tox-pep8/3e94357/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/42/666542/6/check/openstack-tox-py27/5d81b84/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/42/666542/6/check/openstack-tox-py36/4dcee69/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/42/666542/6/check/openstack-tox-py37/f7b2e8b/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/42/666542/6/check/openstack-tox-docs/83e7a55/html/ : SUCCESS in 2m 24s\n- build-openstack-releasenotes http://logs.openstack.org/42/666542/6/check/build-openstack-releasenotes/bd8bbd5/html/ : SUCCESS in 3m 26s\n- openstack-tox-functional http://logs.openstack.org/42/666542/6/check/openstack-tox-functional/110085c/ : SUCCESS in 4m 08s\n- openstack-tox-functional-py36 http://logs.openstack.org/42/666542/6/check/openstack-tox-functional-py36/ccd4bfc/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":6},{"id":"db3fc87139fbc111ff1d58086f30402017e86752","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-27 12:05:53.000000000","message":"Patch Set 6:\n\n(2 comments)\n\noops, I have missed this patch and started work in another place [1]...\n\n@Pawel\n \n\u003e What is your opinion on this?\n\nYeah, that is exactly what I did in [1].\n\n\n[1] https://review.opendev.org/#/c/695865/","accounts_in_message":[],"_revision_number":6},{"id":"de5ba8970f9cd3320f85aa9381b68516653b29ea","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-09 04:59:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6f744aebbae869798172ecf9323391f87fcf9f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 05:10:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae2eb355ca943e0b9a5e11fb27fd1e3 : SUCCESS in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f96067b7a89c4663aa6e63aa42406fa3 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23e57bee6e384fabb5e7ceaa82b91f0e : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffb40585f614258ba2b19a22e3dbc7b : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee24d81d181048d4b50dca9d95c1d398 : SUCCESS in 2m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8009ee4f382b4be9a0c7dbae02902dcc : SUCCESS in 3m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/22fa4a554ee445f089a6e807edebab69 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":7},{"id":"0d6bd7faa78b22d7778838dc734eb322ae3a7ace","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-09 05:36:22.000000000","message":"Patch Set 7:\n\nSquashed [1] to this according to the idea, \"single --member-of option would be enough and less confusing\" that Pawel suggested in the comment on PS6.\n\n\n[1] https://review.opendev.org/#/c/695865/","accounts_in_message":[],"_revision_number":7},{"id":"bd8cf0e47c725d964a3053525c14596e90ec99e9","author":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"date":"2019-12-11 09:22:15.000000000","message":"Patch Set 7: Code-Review+1\n\nIt is exactly what I meant in my previous comment. Thanks for taking care of it.","accounts_in_message":[],"_revision_number":7},{"id":"8d4fb773ec6eedc5e7442aa1cd530d544e2c164a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-13 12:48:41.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ef1270ebe0af2b06bd2c24fadea7ced62865e616","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-14 05:10:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"baaedd9f099819beaa4c5980e3bef78e0182abd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 05:20:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a4ff0f0643451ebd9eb3f6fc367e99 : SUCCESS in 2m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e4fbc85133c4c58b16482f2c870649b : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a63ee83e56ca4fc2a89b895449265678 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cadd711e332146ed9b39bfa745f98549 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664441eecf424535a7653e154e5e3481 : SUCCESS in 3m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ab297c9f43d4c17a337d661defa34ed : SUCCESS in 3m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/75384bb340a74098bf1f39e8ed0069f8 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":8},{"id":"49a17947af1c3d67ba2392dbedfa754b7896e077","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-14 06:12:55.000000000","message":"Patch Set 7:\n\nPS7..8: Trivial rebase for a change to rename the argument [1]\n\n[1] https://review.opendev.org/#/c/695864/5/osc_placement/tests/functional/base.py@83","accounts_in_message":[],"_revision_number":7},{"id":"3b810e9be5b16fb9b53463bf7bcafb22f6182f21","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-14 10:37:08.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0117d36dd006a77cb46cbbfa135cd11da4b6b919","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-15 01:18:08.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"1d80c0111b931ac8c601acfd1b68551ab58e0d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 01:30:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/303ff0dc037c491eabd529b49c8f0873 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44eec948101b4df0aa43bd2e6855e8d2 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3823ebd81261462c817132e9621a8736 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/721e26f5c4894a63a07b0077d2a929b8 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3802c2e57a264372a04a82db334ea5e8 : SUCCESS in 3m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b8be96c74d4462981781f654308aae2 : SUCCESS in 4m 29s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/32388911891646be96c57f361a76424e : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":9},{"id":"820edafe942c75920c5bad0806ba2335ef4e4ce3","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-09-11 06:26:01.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"197ae13cc6dcd0ab0d30389de60c9f2d84ef35bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 07:30:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d307d4b3d94e4c668e8f95fb3dd8181f : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0147e843b20a44fcb3a153d35ea06976 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d29811d0fe1416b8b3ab3afbbaf6adb : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd099245c1964b4ca66ae1ecb5eba5af : SUCCESS in 20m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c5f891b43064735b0064f295a9f4946 : SUCCESS in 4m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/57b04ba988e94910bc38dd6986249c2b : SUCCESS in 5m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1d3ede98f10f4a41be89a0f8e4662fc9 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":10},{"id":"ef4d77073e5e74b09ea1f1506e33c0fecf9355a8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-19 15:03:45.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(5 comments)\n\nAn easy review. I don\u0027t know how I missed this previously.","accounts_in_message":[],"_revision_number":10},{"id":"eff231d24346e5e5ebb65fc991dd16216b0b7674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:06:41.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"2bfa25d212a7ccae073fcea2f6559cd324f9d814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:22:19.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b6e233bb4342af9a9e09362d100c07 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46021b65cceb41839f69d283ab6a9b75 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a7d4464658a4d00970abbcd406da165 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a04f5c30a0402486b494d9100fce03 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1f1e8b1cc524b4d9b48a22871d41d90 : SUCCESS in 4m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1070a084042f4076874d2f93e6350418 : SUCCESS in 5m 51s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6077ac6690e1429e924cded5026b5134 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/49a1b80fe6db427faa9c9a79dd871123 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":10},{"id":"4089a2ae0f6597e3a4fd4a29b2638b5d70d8b384","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:22:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"ae5e8a68a4ffc3ce2e4bbc9c2b2a575144dd7ac5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:28:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d6bb0fa7ca4af3bd2249624f167877 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06cd166a4351494a9ee653fb8a1cc192 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"db98792e466b51ad8112f0c4269ef0298befd50f","revisions":{"890a9c39330909fa7bb905c272774ca2cdabad49":{"kind":"REWORK","_number":1,"created":"2019-06-20 11:27:41.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/42/666542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/1"}}},"commit":{"parents":[{"commit":"e52fa969f9ee7ff2e58ce86c2a6261a80b819093","subject":"Add support for 1.22 microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-20 11:18:09.000000000","tz":120},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-20 11:18:09.000000000","tz":120},"subject":"Add support for 1.24 microversion","message":"Add support for 1.24 microversion\n\nThe 1.24 microversion of placement adds support for listing\nresource providers that are associated with multiple UUIDs.\nUser should use `in:` keyword for every UUID that the requested\nresource provider should be associated with.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/890a9c39330909fa7bb905c272774ca2cdabad49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/890a9c39330909fa7bb905c272774ca2cdabad49"}]},"branch":"refs/heads/master"},"6ebf35962837c1b64e89bbdf4ecad10faafce0ba":{"kind":"REWORK","_number":2,"created":"2019-06-20 11:30:21.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/42/666542/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/2"}}},"commit":{"parents":[{"commit":"e52fa969f9ee7ff2e58ce86c2a6261a80b819093","subject":"Add support for 1.22 microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-20 11:18:09.000000000","tz":120},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-20 11:29:57.000000000","tz":120},"subject":"Add support for 1.24 microversion","message":"Add support for 1.24 microversion\n\nThe 1.24 microversion of placement adds support for listing\nresource providers that are associated with multiple UUIDs.\nUser should use `in:` keyword for every UUID that the requested\nresource provider should be associated with.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6ebf35962837c1b64e89bbdf4ecad10faafce0ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6ebf35962837c1b64e89bbdf4ecad10faafce0ba"}]},"branch":"refs/heads/master"},"34d6631fe2f812f00e3eb69bdfae57e7153411e5":{"kind":"REWORK","_number":3,"created":"2019-06-26 08:03:20.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/42/666542/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/3"}}},"commit":{"parents":[{"commit":"e52fa969f9ee7ff2e58ce86c2a6261a80b819093","subject":"Add support for 1.22 microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-20 11:18:09.000000000","tz":120},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-26 08:00:53.000000000","tz":120},"subject":"Add support for 1.24 microversion","message":"Add support for 1.24 microversion\n\nThe 1.24 microversion of placement adds support for listing\nresource providers that are members of ALL specified resource\nprovider aggregates.\n\nWhen several UUIDs are passed to the single --aggreagate-uuid\noption with \u0027in:\u0027 keyword, the final result will contain resource\nprividers that are associated with ANY of the UUIDs specified in\nthis occurence of the option, AND are also associated with UUIDs\nspecified in other occurences.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/34d6631fe2f812f00e3eb69bdfae57e7153411e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/34d6631fe2f812f00e3eb69bdfae57e7153411e5"}]},"branch":"refs/heads/master"},"bd24fe1191501fe6a11bea7c8d774f5caa674c63":{"kind":"REWORK","_number":4,"created":"2019-06-26 10:09:59.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/42/666542/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/4"}}},"commit":{"parents":[{"commit":"e52fa969f9ee7ff2e58ce86c2a6261a80b819093","subject":"Add support for 1.22 microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-20 11:18:09.000000000","tz":120},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-26 10:08:58.000000000","tz":120},"subject":"Add support for 1.24 microversion","message":"Add support for 1.24 microversion\n\nThe 1.24 microversion of placement adds support for listing\nresource providers that are members of ALL specified resource\nprovider aggregates.\n\nWhen several UUIDs are passed to the single --aggreagate-uuid\noption with \u0027in:\u0027 keyword, the final result will contain resource\nprividers that are associated with ANY of the UUIDs specified in\nthis occurence of the option, AND are also associated with UUIDs\nspecified in other occurences.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bd24fe1191501fe6a11bea7c8d774f5caa674c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bd24fe1191501fe6a11bea7c8d774f5caa674c63"}]},"branch":"refs/heads/master"},"ae9520fae7b7c9e7ebc76322ebc4ed5fa7362076":{"kind":"REWORK","_number":5,"created":"2019-06-26 11:50:55.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/42/666542/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/5"}}},"commit":{"parents":[{"commit":"e52fa969f9ee7ff2e58ce86c2a6261a80b819093","subject":"Add support for 1.22 microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-20 11:18:09.000000000","tz":120},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-26 11:48:54.000000000","tz":120},"subject":"Add support for 1.24 microversion","message":"Add support for 1.24 microversion\n\nThe 1.24 microversion of placement adds support for listing\nresource providers that are members of ALL specified resource\nprovider aggregates.\n\nWhen several UUIDs are passed to the single --aggreagate-uuid\noption with \u0027in:\u0027 keyword, the final result will contain resource\nprividers that are associated with ANY of the UUIDs specified in\nthis occurence of the option, AND are also associated with UUIDs\nspecified in other occurences.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ae9520fae7b7c9e7ebc76322ebc4ed5fa7362076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ae9520fae7b7c9e7ebc76322ebc4ed5fa7362076"}]},"branch":"refs/heads/master"},"2f939338a8ec9db8adad1426559a65ae2df2dfbe":{"kind":"REWORK","_number":6,"created":"2019-07-30 13:32:51.000000000","uploader":{"_account_id":30143,"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","username":"pawel-baclawski"},"ref":"refs/changes/42/666542/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/6"}}},"commit":{"parents":[{"commit":"e52fa969f9ee7ff2e58ce86c2a6261a80b819093","subject":"Add support for 1.22 microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/e52fa969f9ee7ff2e58ce86c2a6261a80b819093"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-06-20 11:18:09.000000000","tz":120},"committer":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-07-30 13:31:39.000000000","tz":120},"subject":"Add support for 1.24 microversion","message":"Add support for 1.24 microversion\n\nIntroduction of --member-of and --member-in options for filtering listed\nresource providers based on their association with other resource\nproviders.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2f939338a8ec9db8adad1426559a65ae2df2dfbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2f939338a8ec9db8adad1426559a65ae2df2dfbe"}]},"branch":"refs/heads/master"},"1dc69487c5275d9844cc8c6b6ce5a02cf5cba142":{"kind":"REWORK","_number":7,"created":"2019-12-09 04:59:52.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/42/666542/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/7"}}},"commit":{"parents":[{"commit":"df07d184ecf8395f2b7f67351b52bf81eb1d4acf","subject":"Improve tests for warning messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/df07d184ecf8395f2b7f67351b52bf81eb1d4acf"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-11-24 08:32:17.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-26 16:57:54.000000000","tz":0},"subject":"Support multiple member_of query parameter","message":"Support multiple member_of query parameter\n\nThe ``openstack allocation candidate list`` and the ``openstack\nallocation candidate list`` command now supports ``--member-of``\noption, which accepts a list of comma-separated UUIDs of the\nresource provider aggregates. If this is specified, the returned\nresource providers must be associated with at least one of the\naggregates identified by uuid. This option can be repeated to\nadd(restrict) the condition with ``--os-placement-api-version 1.24``\nor greater.\n\nThis deprecates the ``--aggregate-uuid`` option because it is\ndifficult to keep using this option to support multiple ``member_of``\nfeature added in microversion 1.24.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\nCo-Autored-By: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nStory: #2006940\nTask: #37613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1dc69487c5275d9844cc8c6b6ce5a02cf5cba142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/1dc69487c5275d9844cc8c6b6ce5a02cf5cba142"}]},"branch":"refs/heads/master"},"249de97de51698481512d587cc0983257040b29d":{"kind":"REWORK","_number":8,"created":"2020-01-14 05:10:33.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/42/666542/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/8"}}},"commit":{"parents":[{"commit":"0c750f450b3b470494dc53db102082bd987e3bc7","subject":"Improve tests for warning messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0c750f450b3b470494dc53db102082bd987e3bc7"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-11-24 08:32:17.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-01-09 11:04:14.000000000","tz":0},"subject":"Support multiple member_of query parameter","message":"Support multiple member_of query parameter\n\nThe ``openstack allocation candidate list`` and the ``openstack\nallocation candidate list`` command now supports ``--member-of``\noption, which accepts a list of comma-separated UUIDs of the\nresource provider aggregates. If this is specified, the returned\nresource providers must be associated with at least one of the\naggregates identified by uuid. This option can be repeated to\nadd(restrict) the condition with ``--os-placement-api-version 1.24``\nor greater.\n\nThis deprecates the ``--aggregate-uuid`` option because it is\ndifficult to keep using this option to support multiple ``member_of``\nfeature added in microversion 1.24.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\nCo-Autored-By: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nStory: #2006940\nTask: #37613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/249de97de51698481512d587cc0983257040b29d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/249de97de51698481512d587cc0983257040b29d"}]},"branch":"refs/heads/master"},"9a69a6a7fd857b90e4fcf5d4c1a404ad0a92b37b":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-01-15 01:18:08.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/42/666542/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/9"}}},"commit":{"parents":[{"commit":"d80651a389072e59cf2aaf80af3209f5674a077b","subject":"Improve tests for warning messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/d80651a389072e59cf2aaf80af3209f5674a077b"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-11-24 08:32:17.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-01-10 07:26:10.000000000","tz":0},"subject":"Support multiple member_of query parameter","message":"Support multiple member_of query parameter\n\nThe ``openstack allocation candidate list`` and the ``openstack\nallocation candidate list`` command now supports ``--member-of``\noption, which accepts a list of comma-separated UUIDs of the\nresource provider aggregates. If this is specified, the returned\nresource providers must be associated with at least one of the\naggregates identified by uuid. This option can be repeated to\nadd(restrict) the condition with ``--os-placement-api-version 1.24``\nor greater.\n\nThis deprecates the ``--aggregate-uuid`` option because it is\ndifficult to keep using this option to support multiple ``member_of``\nfeature added in microversion 1.24.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\nCo-Autored-By: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nStory: #2006940\nTask: #37613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9a69a6a7fd857b90e4fcf5d4c1a404ad0a92b37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9a69a6a7fd857b90e4fcf5d4c1a404ad0a92b37b"}]},"branch":"refs/heads/master"},"db98792e466b51ad8112f0c4269ef0298befd50f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-09-11 06:26:01.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/42/666542/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/42/666542/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/42/666542/10"}}},"commit":{"parents":[{"commit":"25ef9ede49172a47c305e28f6a02d0c0b472920b","subject":"Merge \"tox: Trivial cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/25ef9ede49172a47c305e28f6a02d0c0b472920b"}]}],"author":{"name":"Pawel Baclawski","email":"baclawski.pawel@gmail.com","date":"2019-11-24 08:32:17.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-09-11 06:23:09.000000000","tz":0},"subject":"Support multiple member_of query parameter","message":"Support multiple member_of query parameter\n\nThe ``openstack allocation candidate list`` and the ``openstack\nallocation candidate list`` command now supports ``--member-of``\noption, which accepts a list of comma-separated UUIDs of the\nresource provider aggregates. If this is specified, the returned\nresource providers must be associated with at least one of the\naggregates identified by uuid. This option can be repeated to\nadd(restrict) the condition with ``--os-placement-api-version 1.24``\nor greater.\n\nThis deprecates the ``--aggregate-uuid`` option because it is\ndifficult to keep using this option to support multiple ``member_of``\nfeature added in microversion 1.24.\n\nChange-Id: I0f3068c411c504380918f27f994734528fe55c77\nCo-Autored-By: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nStory: #2006940\nTask: #37613\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/db98792e466b51ad8112f0c4269ef0298befd50f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/db98792e466b51ad8112f0c4269ef0298befd50f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
