)]}'
{"id":"openstack%2Fosc-placement~669478","triplet_id":"openstack%2Fosc-placement~master~I530fcaba664cd21e46ddf8011919954281a04865","project":"openstack/osc-placement","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I530fcaba664cd21e46ddf8011919954281a04865","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 19:43:06.000000000","updated":"2019-08-06 08:46:23.000000000","submitted":"2019-08-06 08:46:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669478-1565081183450-efd57ca1","meta_rev_id":"3b4861a5faaa8de0d511f8f4fd090f1b3644536b","_number":669478,"virtual_id_number":669478,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-08-06 08:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2019-08-06 06:17:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2019-07-31 09:29:59.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2019-08-06 08:35:34.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 06:17:58.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-08-06 08:35:34.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-08-06 08:46:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91214c228daa26036a95704e30e86be0d04b68f6","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 19:43:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3400a3e2c809364ebbce00d2d28e57301777af69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:56:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/78/669478/1/check/test-release-openstack/84505f8/ : SUCCESS in 2m 00s\n- openstack-tox-pep8 http://logs.openstack.org/78/669478/1/check/openstack-tox-pep8/f09a3dc/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/78/669478/1/check/openstack-tox-py27/02ce899/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/78/669478/1/check/openstack-tox-py36/d28adf1/ : SUCCESS in 3m 01s\n- openstack-tox-py37 http://logs.openstack.org/78/669478/1/check/openstack-tox-py37/1f1f963/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/78/669478/1/check/openstack-tox-docs/d4d5636/html/ : SUCCESS in 2m 36s\n- build-openstack-releasenotes http://logs.openstack.org/78/669478/1/check/build-openstack-releasenotes/40e190e/html/ : SUCCESS in 3m 26s\n- openstack-tox-functional http://logs.openstack.org/78/669478/1/check/openstack-tox-functional/26f27ae/ : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/78/669478/1/check/openstack-tox-functional-py36/89c3298/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7cf9cf516615e39c85ec4be7476d1ab1350bd29d","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-31 09:29:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed5bd033ee1cb4f4724719b82e797ad3684c3206","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-08-06 06:17:58.000000000","message":"Patch Set 1: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"8ba0b7ff0b4fbbd821c73413d3e46f654f1d19bc","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-06 08:35:34.000000000","message":"Patch Set 1: Workflow+1\n\nIt\u0027s been long enough and it\u0027s tests, so going to approve without the other +2 because why wait?","accounts_in_message":[],"_revision_number":1},{"id":"3634c4db3e8e2f26d6722a6a10d0affb0c51fc2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 08:35:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa29eeab60d8285b5e93cbeeb9d85efc3cc95faf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 08:46:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2d58763f8a2d72aabdeb121e6e3b0fcd76e6c723","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 08:46:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://logs.opendev.org/78/669478/1/gate/test-release-openstack/3959a73/ : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://logs.opendev.org/78/669478/1/gate/openstack-tox-pep8/504102d/ : SUCCESS in 3m 06s\n- openstack-tox-py27 https://logs.opendev.org/78/669478/1/gate/openstack-tox-py27/a2052b7/ : SUCCESS in 3m 34s\n- openstack-tox-py36 https://logs.opendev.org/78/669478/1/gate/openstack-tox-py36/505a11d/ : SUCCESS in 3m 44s\n- openstack-tox-py37 https://logs.opendev.org/78/669478/1/gate/openstack-tox-py37/b5c6822/ : SUCCESS in 3m 44s\n- openstack-tox-docs https://logs.opendev.org/78/669478/1/gate/openstack-tox-docs/8208e1f/html/ : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://logs.opendev.org/78/669478/1/gate/build-openstack-releasenotes/5ab8d17/html/ : SUCCESS in 2m 41s\n- openstack-tox-functional https://logs.opendev.org/78/669478/1/gate/openstack-tox-functional/12f59e0/ : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://logs.opendev.org/78/669478/1/gate/openstack-tox-functional-py36/209d42d/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75f2480427258f0175fae85627c4e5eb79430ceb","revisions":{"75f2480427258f0175fae85627c4e5eb79430ceb":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:43:06.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/78/669478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/78/669478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/78/669478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/78/669478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/78/669478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/78/669478/1"}}},"commit":{"parents":[{"commit":"78b1f3ea8239ad06b9bf9c9abb78f00707a805a7","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/78b1f3ea8239ad06b9bf9c9abb78f00707a805a7"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:57:40.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:43:06.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I530fcaba664cd21e46ddf8011919954281a04865\nStory: #2005924\nTask: #34236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/75f2480427258f0175fae85627c4e5eb79430ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/75f2480427258f0175fae85627c4e5eb79430ceb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
