)]}'
{"id":"openstack%2Fosc-placement~680295","triplet_id":"openstack%2Fosc-placement~master~I8d2c25a654928d8d46d5e1d0ee203600c9a72600","project":"openstack/osc-placement","branch":"master","topic":"allocation-ratios","hashtags":[],"change_id":"I8d2c25a654928d8d46d5e1d0ee203600c9a72600","subject":"Verify result for inventory set --dry-run","status":"MERGED","created":"2019-09-05 04:37:03.000000000","updated":"2021-05-14 15:22:17.000000000","submitted":"2021-05-14 15:22:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680295-allocation-ratios","meta_rev_id":"18a068910b862c5b2eefde99120305a1e737a164","_number":680295,"virtual_id_number":680295,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 15:22:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-05-14 15:22:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-05-14 15:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 10:01:31.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2020-08-06 15:33:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-09-11 07:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72ee053ed40cf9b15533e28fe54f555737e06dd7","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-05 04:37:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76e83955278824727f6e47303c7ca191e5035049","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 04:47:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d46de62ef1014656a05b9637590eb3b5 : FAILURE in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86ff4cf7fe1e4ddfb56d17e9a5f512f3 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89700ecdc27648d797d88b59c59448ac : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bb2a53757f0e482fa6560551620d90a0 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9cc4d982dce4c3889d350cb617472f7 : SUCCESS in 3m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dfd545196d4543ab8601735ad7f429af : SUCCESS in 4m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b191ae1bb54c4bff806c31e791818d72 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a47e6d805997b7d7dfe48ad249bc9221802ed13d","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-05 05:41:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b8d2c94c4771979afcfdf44b78494ece140ece8","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-09-05 05:47:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a58e5f9e19ecb5e036ab71e1149469666df2444f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 05:58:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cf6d7ff77674071b8a974b3dc0f7ece : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8a94eefbeea4d0a98fb0b74d3b95e65 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42cfce21a1204801a7f255407f9b8897 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/634eb1fd3b7f447f859981268b653776 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2896a42fea644381a26a2743a64d9a3e : SUCCESS in 3m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/521e7560c6d2432796399a48a332ffb5 : SUCCESS in 3m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8c26ea8b194d407db6795b6c3f6a702e : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"7c620f14f3ea83046ec14e6650a8f90a692c5ffa","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-10-21 10:01:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nnit in the commit summary but otherwise seems good","accounts_in_message":[],"_revision_number":3},{"id":"1ad4d6c7f4478e161dea3433c81a620ff3f83352","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-10-25 01:57:14.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d3e262eb8d380416277f527dac6091239397a628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 02:05:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d93375af964a92b9c7063659d93faa : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f50bdeeaf74d4683ac3f40cdd68d25f1 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d797e744156f41cf86b00faaf6c62a5d : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/081c5e86bfb341c49f1eded184be324c : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de20b84714b404e91096baf69c82112 : SUCCESS in 5m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/afb15439244c4667914dee0bad978990 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b9635cf99824475a2e6ea8d7e23051b : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":4},{"id":"3d179f5713a4a849a0907ce183178861b7cb9824","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 15:33:51.000000000","message":"Patch Set 4: Code-Review+2\n\n\\o/","accounts_in_message":[],"_revision_number":4},{"id":"889a3a926f029d373b7bc1f6eda15e819f9ebd45","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-09-11 06:27:18.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"66a5d1941fadd8d471e044befa225b434ad64d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 07:33:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4851b4ea8efa4410bbfda40feaaebd0b : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbe0d887db094dc6b19de7653d48337b : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b28486666c8e4b7b8d1a8517996cb91a : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be98996d3470471eb72e2b946066d6ab : SUCCESS in 18m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/371329b2a18b456d9959ad644ed5815d : SUCCESS in 6m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8aa9f842719d426da6219cc6d6991b6a : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":5},{"id":"170583d81ef0bb3824143a562fb698041f6b67a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2021-05-13 06:29:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"0ce455466b0b767aa1035380a5cb3faea125442a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 06:44:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85bd85714e534775b7f84ad892024c9d : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/854fdc4ccf5e45e4ad7ce666add00e16 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c36b0029bff84bbdbb77903197495d47 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d955f4d053504d468c1c21a4a98ea3af : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e05fce83db4cf5a6b126eaa5be14d7 : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e39906b816e4216a90675c9d42c099b : SUCCESS in 6m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/09187d9ac7094beab7d7bb20cdae78f9 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b291351a367b499e83ba2349fecefc40 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":6},{"id":"236f6c27c3971f01f4e49f722b9dff07fbad2fd3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-14 15:10:02.000000000","message":"Patch Set 6: Workflow+1\n\nThis is test only and has been here for 8 months. Let\u0027s get it in","accounts_in_message":[],"_revision_number":6},{"id":"8ebf1a432bb7c457801907ee1429ee56d5e1656b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:10:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"42bd31866d8aa79c0cad6b67c88338c1f036479e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:22:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b0ee17f2924dbdbc6c66d5233e7917 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e78ee8f800be46ce9922892964472dfc : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c376004b37114d699854fee78fa5a1d3 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/731ee22f967f4d73b69345049c24d8f7 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/893583fec19f47d18c51b089631ab8f4 : SUCCESS in 4m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8d271df531184ff4a50949bd756817e4 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/624fe5cd79d048948fc3e65b22f97409 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":6},{"id":"64519abc25cfeaf29abc06474e8e449d5866167e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:22:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"39befda0d70fe51d75fdece30270fd23b68e0433","revisions":{"ae9012bb8cc5ffb7a7b2c22822fee5103d651764":{"kind":"REWORK","_number":1,"created":"2019-09-05 04:37:03.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/95/680295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/95/680295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/1"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 11:58:48.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 12:14:17.000000000","tz":0},"subject":"Verify result for invenetry set --dry-run","message":"Verify result for invenetry set --dry-run\n\nThere were tests to verify that \u0027resource provider inventory set\u0027\ncommand with --dry-run and --aggregate option changes nothing on\nthe server side but there was no test to verify the result of that\ncommand. This patch adds the test.\n\nChange-Id: I8d2c25a654928d8d46d5e1d0ee203600c9a72600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ae9012bb8cc5ffb7a7b2c22822fee5103d651764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ae9012bb8cc5ffb7a7b2c22822fee5103d651764"}]},"branch":"refs/heads/master"},"433101d1e4fa27595a1fd56e349c0709561d014b":{"kind":"REWORK","_number":2,"created":"2019-09-05 05:41:37.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/95/680295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/95/680295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/2"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 11:58:48.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 13:16:25.000000000","tz":0},"subject":"Verify result for invenetry set --dry-run","message":"Verify result for invenetry set --dry-run\n\nThere were tests to verify that \u0027resource provider inventory set\u0027\ncommand with --dry-run and --aggregate option changes nothing on\nthe server side but there was no test to verify the result of that\ncommand. This patch adds the test.\n\nChange-Id: I8d2c25a654928d8d46d5e1d0ee203600c9a72600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/433101d1e4fa27595a1fd56e349c0709561d014b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/433101d1e4fa27595a1fd56e349c0709561d014b"}]},"branch":"refs/heads/master"},"70b4e0a15d8e3e0aa1b2262b397edfc40fd49875":{"kind":"REWORK","_number":3,"created":"2019-09-05 05:47:18.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/95/680295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/95/680295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/3"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 11:58:48.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 13:23:11.000000000","tz":0},"subject":"Verify result for invenetry set --dry-run","message":"Verify result for invenetry set --dry-run\n\nThere were tests to verify that \u0027resource provider inventory set\u0027\ncommand with --dry-run and --aggregate option changes nothing on\nthe server side but there was no test to verify the result of that\ncommand. This patch adds the test.\n\nChange-Id: I8d2c25a654928d8d46d5e1d0ee203600c9a72600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/70b4e0a15d8e3e0aa1b2262b397edfc40fd49875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/70b4e0a15d8e3e0aa1b2262b397edfc40fd49875"}]},"branch":"refs/heads/master"},"314f2eaff0746601b8d1629ee8d24098a4a8689a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-10-25 01:57:14.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/95/680295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/95/680295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/4"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 11:58:48.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-25 01:56:53.000000000","tz":0},"subject":"Verify result for inventory set --dry-run","message":"Verify result for inventory set --dry-run\n\nThere were tests to verify that \u0027resource provider inventory set\u0027\ncommand with --dry-run and --aggregate option changes nothing on\nthe server side but there was no test to verify the result of that\ncommand. This patch adds the test.\n\nChange-Id: I8d2c25a654928d8d46d5e1d0ee203600c9a72600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/314f2eaff0746601b8d1629ee8d24098a4a8689a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/314f2eaff0746601b8d1629ee8d24098a4a8689a"}]},"branch":"refs/heads/master"},"7cce1ac42afb5944d7c6e5620b2ff4f94759568f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-11 06:27:18.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/95/680295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/95/680295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/5"}}},"commit":{"parents":[{"commit":"25ef9ede49172a47c305e28f6a02d0c0b472920b","subject":"Merge \"tox: Trivial cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/25ef9ede49172a47c305e28f6a02d0c0b472920b"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 11:58:48.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2020-09-11 06:24:36.000000000","tz":0},"subject":"Verify result for inventory set --dry-run","message":"Verify result for inventory set --dry-run\n\nThere were tests to verify that \u0027resource provider inventory set\u0027\ncommand with --dry-run and --aggregate option changes nothing on\nthe server side but there was no test to verify the result of that\ncommand. This patch adds the test.\n\nChange-Id: I8d2c25a654928d8d46d5e1d0ee203600c9a72600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7cce1ac42afb5944d7c6e5620b2ff4f94759568f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/7cce1ac42afb5944d7c6e5620b2ff4f94759568f"}]},"branch":"refs/heads/master"},"39befda0d70fe51d75fdece30270fd23b68e0433":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-05-13 06:29:04.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/95/680295/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/95/680295/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/95/680295/6"}}},"commit":{"parents":[{"commit":"05ca1ee2e16f78524825caf9103205abf6e19497","subject":"Switch default to use latest microversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/05ca1ee2e16f78524825caf9103205abf6e19497"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-09-04 11:58:48.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2021-05-13 06:29:04.000000000","tz":0},"subject":"Verify result for inventory set --dry-run","message":"Verify result for inventory set --dry-run\n\nThere were tests to verify that \u0027resource provider inventory set\u0027\ncommand with --dry-run and --aggregate option changes nothing on\nthe server side but there was no test to verify the result of that\ncommand. This patch adds the test.\n\nChange-Id: I8d2c25a654928d8d46d5e1d0ee203600c9a72600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/39befda0d70fe51d75fdece30270fd23b68e0433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/39befda0d70fe51d75fdece30270fd23b68e0433"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
