)]}'
{"id":"openstack%2Fosc-placement~681231","triplet_id":"openstack%2Fosc-placement~master~Ia00bbf55d8373653d63c9831043c5caaa59b8803","project":"openstack/osc-placement","branch":"master","topic":"cd/readme-bug-link","hashtags":[],"change_id":"Ia00bbf55d8373653d63c9831043c5caaa59b8803","subject":"Update osc-placement bug link in README","status":"MERGED","created":"2019-09-10 12:04:22.000000000","updated":"2019-09-24 14:57:49.000000000","submitted":"2019-09-24 14:55:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681231-1569336957811-d38e73ad","meta_rev_id":"23db6249d5a7541d62469786c7925c3d7bda7719","_number":681231,"virtual_id_number":681231,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 14:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2019-09-20 00:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2019-09-24 14:47:44.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2019-09-24 14:47:44.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 00:49:55.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-09-24 14:47:44.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-24 14:55:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"039a6fe1ab057329d4caec575fa62375b393a577","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-10 12:04:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b882470eef2935ce2ff137769776c6330dcf2c3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:12:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/41ef5d87f8ce4dc8aaadb865f9571636 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c02bf1636a547e99c237ea7102d5bad : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/853d7e9b47be4940ba7155857f66ca87 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"79a106d8bedc7b620e8c033d0e64ab368f1c521e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-10 12:28:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b3a8e1489a03e9b0d8cf399a2cccac7f37ea270c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:38:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/20ccc70bc2554f24962f5ffcede07a32 : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5a5cc821cf4a0f9b8d308140c8793b : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce1ea833ca742db876f1fc20d0daf55 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"061f2b4d770071862cff73b53709a195dc398044","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-09-20 00:49:55.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"686c22d49c164edc12aaa7c2cad9ff73d6a80e8c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 14:47:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nYup matches contributing docs:\n\nhttps://review.opendev.org/#/c/681231/2/CONTRIBUTING.rst\n\nI\u0027m just going to fast approve since this is trivial. It doesn\u0027t appear to show up in the published docs but it\u0027s obviously important for those finding this through github.","accounts_in_message":[],"_revision_number":2},{"id":"af0c2bee999cf93ef12da6df9f7cfa0dc02c2220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 14:47:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec0039c3cfef27ee4743e63983608c9b581b7e40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 14:55:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7083f16c32bd40cb9f27ed7cbcf966a5 : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff3b48662fb44e56967be375f8553d62 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f045d51fe48e415dac9247127163cf25 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"73a1ecb061a82201d99c334a17ba94ef9f3456c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 14:55:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0b7e6d6f6bd11fbf84dda334624fb29bfb64eee4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 14:57:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eabd5663400f4d0d8eaafd8c741e6f36 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"89cdb019715ba15a04a889baf225bfce80c3f866","revisions":{"3497d3f6c6530002401dc9249a22aea370c9aa1e":{"kind":"REWORK","_number":1,"created":"2019-09-10 12:04:22.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/31/681231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/31/681231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/31/681231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/31/681231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/31/681231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/31/681231/1"}}},"commit":{"parents":[{"commit":"bb001133b277e85afba1827f75927ec6d04df683","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bb001133b277e85afba1827f75927ec6d04df683"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-10 12:03:52.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-10 12:04:20.000000000","tz":60},"subject":"Update osc-placement bug link in README","message":"Update osc-placement bug link in README\n\nPoint to storyboard now.\n\nChange-Id: Ia00bbf55d8373653d63c9831043c5caaa59b8803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3497d3f6c6530002401dc9249a22aea370c9aa1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3497d3f6c6530002401dc9249a22aea370c9aa1e"}]},"branch":"refs/heads/master"},"89cdb019715ba15a04a889baf225bfce80c3f866":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-10 12:28:23.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/31/681231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/31/681231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/31/681231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/31/681231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/31/681231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/31/681231/2"}}},"commit":{"parents":[{"commit":"bb001133b277e85afba1827f75927ec6d04df683","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/bb001133b277e85afba1827f75927ec6d04df683"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-10 12:03:52.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-10 12:27:55.000000000","tz":60},"subject":"Update osc-placement bug link in README","message":"Update osc-placement bug link in README\n\nPoint to storyboard now.\n\nChange-Id: Ia00bbf55d8373653d63c9831043c5caaa59b8803\nStory: 2005170\nTask: 29910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/89cdb019715ba15a04a889baf225bfce80c3f866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/89cdb019715ba15a04a889baf225bfce80c3f866"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
