)]}'
{"id":"openstack%2Fosc-placement~681460","triplet_id":"openstack%2Fosc-placement~master~Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9","project":"openstack/osc-placement","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9","subject":"Build pdf docs","status":"MERGED","created":"2019-09-11 12:40:17.000000000","updated":"2019-09-24 15:43:23.000000000","submitted":"2019-09-24 15:41:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681460-1569339717840-f5fe0389","meta_rev_id":"aac4464996a06ef18db6d5de53f0f79a75661810","_number":681460,"virtual_id_number":681460,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 15:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-24 14:58:43.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2019-09-24 15:35:24.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-09-14 15:44:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2019-09-24 14:55:04.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-09-24 15:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2019-09-24 15:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 15:44:49.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-09-24 14:55:04.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-24 15:31:44.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-09-24 15:35:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-24 15:41:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7459830851aa1e33d0e6dd6d9813d8e885bc192","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 12:40:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"377b3f833549a49e8afa82a58484cc8338b11789","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 12:45:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f190b7f9cb065b81535735c4a4bd024abed97047","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 13:01:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7931176e592b4819aa63c7b5a1aaa8ce : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f827eee0eb12420aa0ccf4039b9687a4 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fe46b3c7c7b4439832b811c3f1a4cc3 : SUCCESS in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fd4c6d85d1648f4822eb6c38accfc98 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a62448cff914eccb29a47b0beef8a7f : FAILURE in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf5da7c9cf824a738f281af38057600e : SUCCESS in 3m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d75f53edfcc847a2ac10adb2c36befc3 : SUCCESS in 3m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd6d99198e954154afb6864c8f14f363 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c99e171a61657cb3cf4a1ba11b7a07917bbb366d","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 13:33:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"15cb19ee188aff5381da3a7c5162f1eacf7c1b23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 13:54:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f734853297e14903b043d13114c41d19 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f8554dfda14618b67a6e8e9961c922 : SUCCESS in 2m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5c5d026d143403fa1016a7f24e12c52 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74344b747d2b425b8d8bca44772defac : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5060f40c59bd4520a2332782949ab64b : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf3b8dcff0b49968469d6d1c2b484c4 : FAILURE in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/164be78393a9465c809c0d6ac7fabfc4 : SUCCESS in 5m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0efe475a87b443b39f506753b674f474 : SUCCESS in 4m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e4d62ea387e45e880653fcea2deb5fd : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"eb9b379068f3fbb4424a81e67d2b2171a1eca6c6","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-11 15:21:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cf8afd38cab4ef44df56b014db142d9fb2ed766b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 15:37:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3da69a1a685243e6b27763d84c8278ca : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aff679ca0134aada1f23372b23e7898 : SUCCESS in 2m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82f074dc5227420cb702bb5c636b427a : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89992686d40843fa93922ba667a37d35 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caccde9268744bcc83922e7e7e6a5829 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c9b8222399642fcac09cd8666f17e04 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb375f038fce42ef906a3d0526ddf7a9 : SUCCESS in 3m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/818db3dded1e40f380e9400e6252d7af : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9851c28deca4141a327d9267b1a6ce6 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":4},{"id":"c59e1cedece3bc84f262e832d700bfb004f5d862","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-13 10:31:49.000000000","message":"Topic changed from build-bdf-docs to build-pdf-docs","accounts_in_message":[],"_revision_number":4},{"id":"8da955e9a1a6908841362df9e5a4dc60c3865347","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-09-14 15:44:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"020007fb7af931cd33889af6d503539ac4f18b19","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 14:55:04.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nQuestion inline about the badge thing but this does generate a pdf:\n\nhttps://df8f450b5ac356fa04e7-6f50b24062b4b6807f764f20ef239d75.ssl.cf1.rackcdn.com/681460/4/check/openstack-tox-docs/2c9b822/pdf/doc-osc-placement.pdf\n\nSome of the examples in the docs are rendered a bit weirdly in the pdf but beggars can\u0027t be choosers at this point in the release.","accounts_in_message":[],"_revision_number":4},{"id":"859f7e418e339d7f0f4c154c2a59568274d64cb6","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 14:58:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5a5141191629c7a5b966113d21e6248e152062e9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 15:08:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"364f3d0b19ce030172bbf719077a9bb5c787ec86","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-09-24 15:31:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM. I would add the necessary bindeps to the osc-placement bindep.txt in a followup.","accounts_in_message":[],"_revision_number":4},{"id":"f628601e75dbcfb4fb6fafcfc15739e8c2370659","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:31:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"78d07e3f2754bb7c6d01677a467a44a3f704fc87","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 15:35:24.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3b15b8ce7c8b5199d12dd018a7e261d2d47a9f4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:41:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2528d86a0a7878ea73f960f9ecceca8cf92b4a34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:41:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a6f54646177496f88f78697c93501f4 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86cc0e7e370f4e5798c6b456f0abd6ce : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5f052ad40684df5a6100fad9a6a2813 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cb909e24f76482592c46e64aa42cfe6 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47c1c7aec3e84237ae4f98b76dcea329 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95dc9e31adf2494c8b9e315fd426ed7c : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8e273039d6641e6877c4ae6577d31a1 : SUCCESS in 3m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4635c9763325457b822110ced0b6f224 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/093593c85db248f89266483fd23f94c3 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":4},{"id":"aac4464996a06ef18db6d5de53f0f79a75661810","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:43:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3924096c195548b5a4545d09752d39b0 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b908b3e516af4f438fd8e59bff63d375 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"99d879abfd73031ea8481eb043e064b27db968ba","revisions":{"26bbe0a2b6d1ff36f947799af507d4e50c7aa9eb":{"kind":"REWORK","_number":1,"created":"2019-09-11 12:40:17.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/60/681460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/60/681460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/1"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:38:50.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:38:50.000000000","tz":60},"subject":"WIP: Build pdf docs","message":"WIP: Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/26bbe0a2b6d1ff36f947799af507d4e50c7aa9eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/26bbe0a2b6d1ff36f947799af507d4e50c7aa9eb"}]},"branch":"refs/heads/master"},"744bee106f6ec9ed5c42fea40621b12bcf38beb6":{"kind":"REWORK","_number":2,"created":"2019-09-11 12:45:28.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/60/681460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/60/681460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/2"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:38:50.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:45:12.000000000","tz":60},"subject":"WIP: Build pdf docs","message":"WIP: Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/744bee106f6ec9ed5c42fea40621b12bcf38beb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/744bee106f6ec9ed5c42fea40621b12bcf38beb6"}]},"branch":"refs/heads/master"},"b94604186d450ecf55d707def9c17b3efcdfbb57":{"kind":"REWORK","_number":3,"created":"2019-09-11 13:33:50.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/60/681460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/60/681460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/3"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:38:50.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 13:30:38.000000000","tz":60},"subject":"WIP: Build pdf docs","message":"WIP: Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\nTo get a build, the \"badge\" in the README has been removed.\nThe other option to add some svg -\u003e pdf support, but that seems\noverkill for a badge which is effectively meaningless.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b94604186d450ecf55d707def9c17b3efcdfbb57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/b94604186d450ecf55d707def9c17b3efcdfbb57"}]},"branch":"refs/heads/master"},"99d879abfd73031ea8481eb043e064b27db968ba":{"kind":"REWORK","_number":4,"created":"2019-09-11 15:21:49.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/60/681460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/60/681460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/60/681460/4"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 12:38:50.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-11 15:21:06.000000000","tz":60},"subject":"Build pdf docs","message":"Build pdf docs\n\nThis follows the instructions [1] in an attempt to build pdf docs.\n\nTo get a build, the \"badge\" in the README has been removed.\nThe other option to add some svg -\u003e pdf support, but that seems\noverkill for a badge which is effectively meaningless.\n\nThe build wanted \u0027xindy\u0027 to build an index, but we can turn that\noff and still get useful results.\n\n[1] https://etherpad.openstack.org/p/train-pdf-support-goal\n\nChange-Id: Icf7c22bf9d1de6fb2a74a756c370930d4c00b0b9\nStory: 2006110\nTask: 35397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/99d879abfd73031ea8481eb043e064b27db968ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/99d879abfd73031ea8481eb043e064b27db968ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
