)]}'
{"id":"openstack%2Fosc-placement~683941","triplet_id":"openstack%2Fosc-placement~master~Id4f8f0224f61c1741e6a1a8369ec053bb1824e8a","project":"openstack/osc-placement","branch":"master","topic":"cd/os-endpoint","hashtags":[],"change_id":"Id4f8f0224f61c1741e6a1a8369ec053bb1824e8a","subject":"Use os-endpoint instead of os-url for functional tests","status":"MERGED","created":"2019-09-23 12:14:49.000000000","updated":"2019-09-24 12:07:23.000000000","submitted":"2019-09-24 11:12:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683941-1569323534214-c425b5ec","meta_rev_id":"d98d579a30517a1f4fd995671cebba876d460c0f","_number":683941,"virtual_id_number":683941,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 11:12:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2019-09-24 11:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":1,"date":"2019-09-24 11:02:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 11:02:29.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-09-24 11:12:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abc936c291047d4e000bcfee0436429fe2caf819","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-23 12:14:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"787edc86c94e09ee2ca24117486d4e2a71bd8264","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 12:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31f27a04dad34345a4146e3a6bc79339 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e175a2a391924b1d8161d993dae44406 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db7ee6fdeb324aaeb02effb75178d623 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ef894711ae94fa9ab32e508235089d7 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f27a40feb82f454491a46cec4cf4037e : SUCCESS in 4m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2e8e6e570848467b8f9f7e9602b09618 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63016b33b32447b3b0f2ce0c4dbd324b : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4fbdc6660a80ec5b101b3eb09859375f3c942b56","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-09-24 11:00:43.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"984e81fee2f37e4faa80e2fa3c5e6ea64458d711","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-09-24 11:02:29.000000000","message":"Patch Set 1: Workflow+1\n\nApproving it now to unblock the gate.","accounts_in_message":[],"_revision_number":1},{"id":"f185f5719cd01dc2bc89044594e4ac3931963994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 11:02:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a479b95f59f085ba88d0fd784e4096dff6ea8970","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 11:12:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c84e2eee251baf53a1e3d9522f097e1d739009f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 11:12:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03db9ba300074e7c932c4a681e7d4446 : SUCCESS in 2m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d07b8320d3d4cb28dea1da188f2d3a6 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0778d425991c46c097e6e838729b3265 : SUCCESS in 2m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdb1435950324b468075f9d5263ba323 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bbae12e83ff4604a582f39ae86527e0 : SUCCESS in 3m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/70d6fa51c4164db983b73fb37b22db92 : SUCCESS in 3m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/baa4ab577a5741a19caeea8b4f0b05b2 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"122f6288fd23df516c7cb517c53f01b8b8f8f1e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 11:13:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9cffb3d82d845e5b1bcf5751ee2c3a1 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"159daaa00a1d67a0326a525cf36e4c2cf94868ad","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 12:07:23.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 878fa6112947d2eb022937af11dec5f415d9fa57","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f49b9fcc81918ca937ed7332702432c3f82eb639","revisions":{"f49b9fcc81918ca937ed7332702432c3f82eb639":{"kind":"REWORK","_number":1,"created":"2019-09-23 12:14:49.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/41/683941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/41/683941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/41/683941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/41/683941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/41/683941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/41/683941/1"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-23 12:13:15.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-23 12:13:15.000000000","tz":60},"subject":"Use os-endpoint instead of os-url for functional tests","message":"Use os-endpoint instead of os-url for functional tests\n\nThe functional tests use an explicit endpoint and token for auth\nto the placement fixture. Before openstack client 4.0.0 os-url\nwas a way to specify the endpoint. Now os-endpoint is the way.\n\nUpdate accordingly.\n\nNote that this doesn\u0027t impact the behavior of the plugin itself,\njust the tests.\n\nChange-Id: Id4f8f0224f61c1741e6a1a8369ec053bb1824e8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f49b9fcc81918ca937ed7332702432c3f82eb639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f49b9fcc81918ca937ed7332702432c3f82eb639"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
