)]}'
{"id":"openstack%2Fosc-placement~684292","triplet_id":"openstack%2Fosc-placement~stable%2Ftrain~Id4f8f0224f61c1741e6a1a8369ec053bb1824e8a","project":"openstack/osc-placement","branch":"stable/train","topic":"cd/os-endpoint-stable/train","hashtags":[],"change_id":"Id4f8f0224f61c1741e6a1a8369ec053bb1824e8a","subject":"Use os-endpoint instead of os-url for functional tests","status":"MERGED","created":"2019-09-24 12:07:23.000000000","updated":"2019-09-27 15:33:30.000000000","submitted":"2019-09-27 15:31:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684292-1569598308900-d890bd4f","meta_rev_id":"0fe9ebd961c5ad4593c0510ebd1bb383918f17b3","_number":684292,"virtual_id_number":684292,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2019-09-24 18:30:25.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 15:31:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-27 15:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-24 19:01:00.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-09-24 12:09:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-27 15:21:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 12:09:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-09-24 19:01:00.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-27 15:21:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-27 15:31:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4cd22e66a986ba6f2df513b5f4bab361825c5f3","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 12:07:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"26f4c49fe70a40459cfc9e9ee4ba9157876bd5fa","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-09-24 12:09:47.000000000","message":"Patch Set 1: Code-Review+1\n\nclean backport","accounts_in_message":[],"_revision_number":1},{"id":"2b0c686fc69f99afbb15881ce9686141c56410d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 12:21:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78804752944f49489d617af80cec3534 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3da8c0c64c954bf2b0d62afb7818c470 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8116bc921e44c1f868888b09c5346fa : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a18bea7b830a48a993938e6910d41111 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e878ea872ec24df38f09dc5cb07f1d76 : SUCCESS in 2m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/033b8a58163a4de195276e2232fb0316 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9392eefaf5ec462c9ec34296bdb5d527 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dba4923c84c59e8d9240586daf970f0f16987166","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 14:05:27.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nWhat happens if you have python-openstackclient\u003c4.0 installed? Does this still work?","accounts_in_message":[],"_revision_number":1},{"id":"85c64364611804ddbc5b961cbaad858463e01b35","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 14:12:03.000000000","message":"Patch Set 1: Code-Review-1\n\nSounds like this breaks if you have osc\u003c4:\n\n(9:11:39 AM) mriedem: cdent: so does it break if you have osc\u003c4?\n(9:11:45 AM) cdent: it does yeah","accounts_in_message":[],"_revision_number":1},{"id":"e137c537b255065d8406168354832bcb881596c5","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 14:29:00.000000000","message":"Patch Set 1:\n\none option for making this safe is also backporting: https://review.opendev.org/684352\n\nbut that may count as a requirements change (despite being in test-requirements only) so will leave this for further discussion...","accounts_in_message":[],"_revision_number":1},{"id":"5485f7a40f6aa4ad8b7ef993640bc9eb68d971c1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 14:38:02.000000000","message":"Patch Set 1:\n\nBumping a test requirement isn\u0027t the end of the world on stable, but it\u0027d be nice to know why the osc\u003c4 code doesn\u0027t handle --os-endpoint since it seems to be advertised that it should.","accounts_in_message":[],"_revision_number":1},{"id":"6b0bcf8db7720514e1ce5b681f98d89f873766f8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 15:29:17.000000000","message":"Patch Set 1:\n\nOnce we cherry-pick https://review.opendev.org/#/c/684370/ on top of this change we can land both.","accounts_in_message":[],"_revision_number":1},{"id":"e0de3a3f797009845e53bd89056004fd29d1115c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 16:01:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6b1d79233c90af67e0dc7c958dfe0b4f4807dd3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 18:21:51.000000000","message":"Patch Set 1: -Code-Review\n\nI\u0027m not even sure that we need this if we have https://review.opendev.org/#/c/684392/ ?","accounts_in_message":[],"_revision_number":1},{"id":"55db6abe9a0f5ce866444658458bbc0382fcb226","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 18:30:25.000000000","message":"Patch Set 1:\n\nos-url doesn\u0027t work/exist in 4.0.0.\n\nSince we\u0027re unpinned we need to support the version that works in both old and new osc.","accounts_in_message":[],"_revision_number":1},{"id":"1eab383022597953d9136c40c6c3801b983f56e6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 19:01:00.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e Since we\u0027re unpinned we need to support the version that works in both old and new osc.\n\nGood point.","accounts_in_message":[],"_revision_number":1},{"id":"02dc53a2a601685b38d811d3bf9a8d9621f4d9c9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-27 15:21:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87d7448de66abafaada8c07b76251d1f0b3b56b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:21:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f42574ff731cf4e75e1d1d9f882597d4e25ff71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:31:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"83a91436c43c008ee0b622cf9b5dd3546eaab6dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:31:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e36eaffe68bd4e588b6ce7712cbe6366 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f87a0cd75d74494b2996aa02804e52d : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f1af7c1f624110abf23f031fa064a2 : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c2c4ceac05049d5a1039cfe601f2ecb : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b0d5fc4364400c80db9ad1af5d848e : SUCCESS in 3m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1d2a34f72b8547a5831c9762f52c8d69 : SUCCESS in 3m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/37ec9290f62c4c5cb2875dac143f54ef : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"26c781383558a252a22fe29730fdfdba2f6d4498","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:33:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/496a66bc0a9948c1b01a6b0c89345df5 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"878fa6112947d2eb022937af11dec5f415d9fa57","revisions":{"878fa6112947d2eb022937af11dec5f415d9fa57":{"kind":"REWORK","_number":1,"created":"2019-09-24 12:07:23.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/92/684292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/92/684292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/92/684292/1"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-23 12:13:15.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-24 12:07:23.000000000","tz":0},"subject":"Use os-endpoint instead of os-url for functional tests","message":"Use os-endpoint instead of os-url for functional tests\n\nThe functional tests use an explicit endpoint and token for auth\nto the placement fixture. Before openstack client 4.0.0 os-url\nwas a way to specify the endpoint. Now os-endpoint is the way.\n\nUpdate accordingly.\n\nNote that this doesn\u0027t impact the behavior of the plugin itself,\njust the tests.\n\nChange-Id: Id4f8f0224f61c1741e6a1a8369ec053bb1824e8a\n(cherry picked from commit f49b9fcc81918ca937ed7332702432c3f82eb639)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/878fa6112947d2eb022937af11dec5f415d9fa57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/878fa6112947d2eb022937af11dec5f415d9fa57"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
