)]}'
{"id":"openstack%2Fosc-placement~684352","triplet_id":"openstack%2Fosc-placement~master~I5611b797a5115c7698f208b8a7e4b84bf0dc39f3","project":"openstack/osc-placement","branch":"master","topic":"cd/pin-osc","hashtags":[],"change_id":"I5611b797a5115c7698f208b8a7e4b84bf0dc39f3","subject":"Pin python-openstackclient \u003e\u003d 4.0.0 in test-requirements","status":"ABANDONED","created":"2019-09-24 14:21:59.000000000","updated":"2019-09-24 15:12:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7eed03888278ebb5c7356e617fc31103e89ace3","_number":684352,"virtual_id_number":684352,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-09-24 15:02:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2019-09-24 15:10:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 15:02:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-24 15:10:42.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"}],"messages":[{"id":"65049a7d996d62dd39cb8524b92893209ab108d4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 14:21:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7bddfaf36411a980deeee1840641948b3ac2ab0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 14:24:16.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm, based on this change in osc 4.0.0:\n\nhttps://opendev.org/openstack/python-openstackclient/commit/6fcc2608b17d84cf3699bb4a5bae692404393ca1","accounts_in_message":[],"_revision_number":1},{"id":"7d888a5646a94ce28ba4511cb1686d6d7e189bc6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 14:45:13.000000000","message":"Patch Set 1: -Code-Review\n\nIt sounds like we maybe don\u0027t need this (or at least don\u0027t need to backport this) if you specify \"--os-auth-type token_endpoint\" which is the default with --os-endpoint in osc 4.0 but was not a default in osc \u003c 4.0.","accounts_in_message":[],"_revision_number":1},{"id":"52f68f6bb67b8de271919789ce7e47ddc87a695e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:02:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1c114344a347a094a571a12d91f6e9 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c3420cc4119456987b83a32c8bd546c : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b4529bb18b44afdbdfd3c8ca8872847 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/858115aff93e481e8cf2165ba6f689cd : SUCCESS in 4m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4fbf83f9d80b45e0bb140943826b5e4f : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6189a128923a491db693dc8dc419661f : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8958b559aa7c416ab24275fffa26d0c6 : SUCCESS in 5m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9e0f692f708144dd8ae7ca5ddd010738 : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87f1c7be65a246e182f82fbfabe05ec3 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3d803fd5617e62564edee7c0f5d86c92f9dab5c1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 15:10:42.000000000","message":"Patch Set 1:\n\nDo you still want to do this even though we have https://review.opendev.org/#/c/684370/ ?","accounts_in_message":[],"_revision_number":1},{"id":"338a6fe2e81d8d04a14ec4568dd069a36653e1e0","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 15:12:49.000000000","message":"Abandoned\n\nSince we now have https://review.opendev.org/#/c/684370/ we don\u0027t need this","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b94467cb923093944b381e760e843e94cb38588","revisions":{"3b94467cb923093944b381e760e843e94cb38588":{"kind":"REWORK","_number":1,"created":"2019-09-24 14:21:59.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/52/684352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/52/684352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/52/684352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/52/684352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/52/684352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/52/684352/1"}}},"commit":{"parents":[{"commit":"f49b9fcc81918ca937ed7332702432c3f82eb639","subject":"Use os-endpoint instead of os-url for functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f49b9fcc81918ca937ed7332702432c3f82eb639"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-24 14:20:32.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-24 14:20:32.000000000","tz":60},"subject":"Pin python-openstackclient \u003e\u003d 4.0.0 in test-requirements","message":"Pin python-openstackclient \u003e\u003d 4.0.0 in test-requirements\n\nIn change Id4f8f0224f61c1741e6a1a8369ec053bb1824e8a we added\ntest functionality to use os-endpoint arg instead of os-url.\n\nThis uses 4.0.0 functionality, so let\u0027s make it explicit that\nwe are using that.\n\nChange-Id: I5611b797a5115c7698f208b8a7e4b84bf0dc39f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3b94467cb923093944b381e760e843e94cb38588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3b94467cb923093944b381e760e843e94cb38588"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
