)]}'
{"id":"openstack%2Fosc-placement~684370","triplet_id":"openstack%2Fosc-placement~master~Ie9c76a23658b46997dcb17ba3fb7f1e700eca506","project":"openstack/osc-placement","branch":"master","topic":"cd/none-auth-type","hashtags":[],"change_id":"Ie9c76a23658b46997dcb17ba3fb7f1e700eca506","subject":"Be explicit about auth type in functional tests","status":"MERGED","created":"2019-09-24 15:00:48.000000000","updated":"2019-09-24 16:00:30.000000000","submitted":"2019-09-24 15:47:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684370-1569340076068-c425b5ec","meta_rev_id":"a4b08b66d123fcd09c7f556196df45f787342bc2","_number":684370,"virtual_id_number":684370,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 15:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-09-24 15:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-09-24 15:10:18.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-09-24 15:36:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 15:10:18.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-24 15:36:05.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-24 15:47:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d79382864a7838d929a96575c43dd22019071ae4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 15:00:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e19a433bb5ab87824026a95c165c135acf783f49","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 15:10:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"81c7dec73d39644744df898f5d64593fd4c9eab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:10:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c89111377fd48fe8a98242f63816f22 : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc28f688e7204b48946edca627869ed9 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3415f90ebb8241cca5aa0a519f27b59d : SUCCESS in 2m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17b0b54e5dcd4dd5bb56dba07501705b : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6631aed1e52c45009402682dfda423b5 : SUCCESS in 4m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9bd3c82ffb3d4a29ada7e57fb8ca1bea : SUCCESS in 4m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a16bef1f79c2431c99a2371aafe8a675 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9c1941662f54334f695cb92ca39212c4fe042f7e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 15:36:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f32575b78aaa9fdd4b7c8f226d7a205433fe7b0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:36:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d6a142f8738a2e3f1ef7d869164d503bfe0abc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:47:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8f660a921e43209cc0913af98feb0e : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a6948af342c42d78b800b6dc81f585b : SUCCESS in 2m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b2ee262d87f4bb0bcd62af54c7d939d : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/316875f1e62748c7abcba66de64992c4 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d696a670c7f4bb39e6db5d15089124f : SUCCESS in 6m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9d0a2e8b55d2415f9525e0304fcb76c0 : SUCCESS in 4m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18fe9f516e784267a65a568c4beca273 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b51931c8e6439f9f9ce92d0c68a7eec86c37680a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:47:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"50ba0dd00334e191687d3417acf7b3e2682cbb12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:49:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/700084ed5c2440939cceeb1632088a22 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e80879487a53baf102425d3b76e05d9c6a4b23af","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 16:00:30.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit ee8875d64bc7046b13696a90ef7a4aa54f83103c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"306a91e9e8939d75e0309e492502fa0805cc2c8d","revisions":{"306a91e9e8939d75e0309e492502fa0805cc2c8d":{"kind":"REWORK","_number":1,"created":"2019-09-24 15:00:48.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/70/684370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/70/684370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/70/684370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/70/684370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/70/684370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/70/684370/1"}}},"commit":{"parents":[{"commit":"f49b9fcc81918ca937ed7332702432c3f82eb639","subject":"Use os-endpoint instead of os-url for functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/f49b9fcc81918ca937ed7332702432c3f82eb639"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-24 14:55:14.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-24 15:00:29.000000000","tz":60},"subject":"Be explicit about auth type in functional tests","message":"Be explicit about auth type in functional tests\n\nWhen using the old style of \u0027--os-url\u0027 we didn\u0027t need to be\nexplicit, it deduced the right type. And python-openstackclient\n\u003e\u003d 4.0.0 it does for --os-endpoint. But when using \u0027--os-endpoint\u0027\nwith \u003c4.0.0 we need to state \u0027admin_token\u0027 explicitly.\n\nThis form works in both 3.19.x and 4.0.x.\n\nChange-Id: Ie9c76a23658b46997dcb17ba3fb7f1e700eca506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/306a91e9e8939d75e0309e492502fa0805cc2c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/306a91e9e8939d75e0309e492502fa0805cc2c8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
