)]}'
{"id":"openstack%2Fosc-placement~684389","triplet_id":"openstack%2Fosc-placement~master~Ib3db3888d0bb1c47773e3e44919c7077d0906ceb","project":"openstack/osc-placement","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Ib3db3888d0bb1c47773e3e44919c7077d0906ceb","subject":"Fixups for pdf docs","status":"MERGED","created":"2019-09-24 15:52:50.000000000","updated":"2019-09-25 07:52:21.000000000","submitted":"2019-09-25 07:50:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684389-1569397836213-9b36cc3c","meta_rev_id":"1dd8b69da9e2f8f31e8b34fe7dd9aab6c16e4884","_number":684389,"virtual_id_number":684389,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 07:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-09-24 16:29:06.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-09-25 07:40:44.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-09-25 07:40:44.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 16:29:06.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-25 07:40:44.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-09-25 07:50:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf3633b2953a4aaf947c6bd8e3ff9469d7173477","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 15:52:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5e69da16a75a275c88586b77479821c23382e59","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 15:55:58.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"4140e2a41480b6f00b93c8073b3aedf7080769d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 16:04:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/41b74347c89b4f909769c15d06882f28 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a3ecebe3354386a2e68bdf491186ec : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad97448f21c24ba19b459cb078512ef3 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ea72d790f239c421ded82e07747575a08fd7c4e4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 16:20:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8e2d1d33309cb3b5c6771251c6300e673f16643b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 16:27:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc63e28d284a7e14d96cc0555e3c677ed3e29009","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 16:27:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"773db6348c2be47e970bbb7e9d1bea9edb0e7792","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 16:29:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5e8942f69da1f7b04c61c0cb63effca9e9f58930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 16:41:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f4f0081215894286be367ff824a779b5 : SUCCESS in 1m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9abb3e100042069dcd8e49484a5202 : SUCCESS in 8m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff3e33c59d99414ba90b3d9d4d7146d9 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6554137cd2414199ac5d5fbeaddf9b15 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0dce21fce58047a7804487854ddfb40a : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/157dbd13bd5a4581855a8bf8914c6d8f : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94e3d71ea5664b069999d2ae81bb95aa : SUCCESS in 6m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1365fdd509424176b08bc788f3f90e87 : SUCCESS in 3m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9db2ee59e62146dcbcc1d02310247e45 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2cdf9735efac97ac8f0b4d6b8b4c1405f6b0ad17","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-25 07:40:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"28f58f16381d761f0cb6f3e075491c63d58a620c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 07:40:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"959e588418e8a43c26eee6fe6d32fc43a29f6cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 07:50:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8fd37f0dbc9c4b6f8a730af10721e9ca : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a9076d71064e348ffe3867f25d9a96 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af81ae5c990d4ccdbc65df962adeaee3 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa22a92586574c3f9ef16da10eab9858 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23828bbfe5fd449e9d6cd67bb1f985f5 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11b881babaf42e9b4232eb31cf4b74c : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc2954d533ec4dbd98a732c5621c40ad : SUCCESS in 4m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e846fcaf21d1426fab9b9392b70149b3 : SUCCESS in 3m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24af1085d30c41a4b4d591b555c46e5c : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a40f04f9c51ac231da1ce72c855b317296b6b6e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 07:50:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a9f86fb0b4dc1eef8c65539d7ea1a06af3fde23a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 07:52:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86dee2b5de247d4b4cc5e3b86c02dc5 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c069bc88c904dcebb4eddc468988cca : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cae0bc84e5fd1ab7878c336bd708e60eeb7e893","revisions":{"2e87576d6dd2ad07b511471eb6fca07d2aa7beb8":{"kind":"REWORK","_number":1,"created":"2019-09-24 15:52:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/684389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/89/684389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/89/684389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/89/684389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/89/684389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/89/684389/1"}}},"commit":{"parents":[{"commit":"2ffbba144cc5d48aa422efa72dceb25e1c320578","subject":"Merge \"Build pdf docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2ffbba144cc5d48aa422efa72dceb25e1c320578"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-24 15:47:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-24 15:47:20.000000000","tz":60},"subject":"Fixups for pdf docs","message":"Fixups for pdf docs\n\n- Don\u0027t remove badges - simply exclude them from the main docs.\n- Don\u0027t include stuff that only makes sense in HTML - namely the indexes\n\nChange-Id: Ib3db3888d0bb1c47773e3e44919c7077d0906ceb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2e87576d6dd2ad07b511471eb6fca07d2aa7beb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2e87576d6dd2ad07b511471eb6fca07d2aa7beb8"}]},"branch":"refs/heads/master"},"0cae0bc84e5fd1ab7878c336bd708e60eeb7e893":{"kind":"REWORK","_number":2,"created":"2019-09-24 16:27:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/684389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/89/684389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/89/684389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/89/684389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/89/684389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/89/684389/2"}}},"commit":{"parents":[{"commit":"2ffbba144cc5d48aa422efa72dceb25e1c320578","subject":"Merge \"Build pdf docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/2ffbba144cc5d48aa422efa72dceb25e1c320578"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-24 15:47:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-09-24 16:26:39.000000000","tz":60},"subject":"Fixups for pdf docs","message":"Fixups for pdf docs\n\n- Don\u0027t remove badges - simply exclude them from the main docs.\n- Don\u0027t include stuff that only makes sense in HTML - namely the indexes\n- Don\u0027t explicitly set basepython in a tox testenv when we can inherit\n  it\n\nChange-Id: Ib3db3888d0bb1c47773e3e44919c7077d0906ceb\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0cae0bc84e5fd1ab7878c336bd708e60eeb7e893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0cae0bc84e5fd1ab7878c336bd708e60eeb7e893"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
