)]}'
{"id":"openstack%2Fosc-placement~684392","triplet_id":"openstack%2Fosc-placement~stable%2Ftrain~Ie9c76a23658b46997dcb17ba3fb7f1e700eca506","project":"openstack/osc-placement","branch":"stable/train","topic":"cd/none-auth-type","hashtags":[],"change_id":"Ie9c76a23658b46997dcb17ba3fb7f1e700eca506","subject":"Be explicit about auth type in functional tests","status":"MERGED","created":"2019-09-24 16:00:30.000000000","updated":"2019-10-02 22:26:56.000000000","submitted":"2019-10-02 22:25:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684392-1570055142952-d890bd4f","meta_rev_id":"6a98cad4ef939058e822c3e0098d1e5e1198cfa3","_number":684392,"virtual_id_number":684392,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 22:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-27 15:20:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-30 05:51:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2019-10-02 22:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-09-25 07:39:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2019-10-02 22:14:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 07:39:56.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-09-27 15:20:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-30 05:51:59.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-10-02 22:14:37.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-02 22:25:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c74678d88080ec8e961ea04501cef11c30732f61","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 16:00:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5eb08d064fd142b082198d42bee25316f69e1943","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 16:00:36.000000000","message":"Topic changed from cd/none-auth-type-stable/train to cd/none-auth-type","accounts_in_message":[],"_revision_number":1},{"id":"9f90e7281263d506da08a2ad73a28ede8516683a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-24 16:00:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b5ef7e62ba5ac937fa389088e3ede054449c5f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 16:12:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3510391c30428d9fc5ecaf2cab049d : SUCCESS in 2m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de9a82934950494eb92cfd0819bedfd3 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dea4cf619b04314a796eeac76c2392d : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/507b10da7f2b426c81fd252754ae581b : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19016b93ed5947a1a6387c97f700d79c : SUCCESS in 4m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6e02081ffb3948cf9af18801eb990d3f : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95dd3b7012444c6094c8c9d19be76c50 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1ae92ac19929c59a3793111904ecaf32865054d1","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-25 07:39:56.000000000","message":"Patch Set 2: Code-Review+1\n\ntidy clean backport\n\n(only +1 since it is my code)","accounts_in_message":[],"_revision_number":2},{"id":"6577f10058ff886b243dc223743607cbef40d863","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-27 15:20:33.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks like a good test update.","accounts_in_message":[],"_revision_number":2},{"id":"5af54cc8f811463bcaff61770eeeafc0280db1f2","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-09-30 05:51:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9e0fe19db6feac6ff1f2319456fb8dba4d770f4c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-02 22:14:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nClean cherry pick, lgtm.","accounts_in_message":[],"_revision_number":2},{"id":"2054c5fff9affdfc567687d7c44dd951bdc898ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 22:14:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04cebc6c5cffbc0fe49f61b3b1b932f63f291db4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 22:25:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4b965a8f1c4a3225d8cc385e5a5bb3a08477002e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 22:25:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91cb24caa274604b055013bad7e8461 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/261a2c28eb604e3aac4cc2cbb0cfefb9 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be4ed4d4f06d44c094243acc403d8181 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08fc018d7800422d8934d5f00d2d19ea : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371d2d85ab9741a5bc89cf0ebe5fb04d : SUCCESS in 3m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c422468c23074d9b9356423df993dd00 : SUCCESS in 4m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1fa8764e7e445b5b990c2668adb64b8 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6a98cad4ef939058e822c3e0098d1e5e1198cfa3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 22:26:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c015f2f685c54e78ab440cd58b04f6e5 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca32d2146a976237cbd32be1ff755c4672f2d9df","revisions":{"ee8875d64bc7046b13696a90ef7a4aa54f83103c":{"kind":"REWORK","_number":1,"created":"2019-09-24 16:00:30.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/92/684392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/92/684392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/92/684392/1"}}},"commit":{"parents":[{"commit":"8bbca01dbddb1eb374d708553ccaddb4f130116e","subject":"Follow up for Ib0cbb58d0adbbcfe83ee48d2ff6c9af1a516a7ae","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/8bbca01dbddb1eb374d708553ccaddb4f130116e"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-24 14:55:14.000000000","tz":60},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-24 16:00:30.000000000","tz":0},"subject":"Be explicit about auth type in functional tests","message":"Be explicit about auth type in functional tests\n\nWhen using the old style of \u0027--os-url\u0027 we didn\u0027t need to be\nexplicit, it deduced the right type. And python-openstackclient\n\u003e\u003d 4.0.0 it does for --os-endpoint. But when using \u0027--os-endpoint\u0027\nwith \u003c4.0.0 we need to state \u0027admin_token\u0027 explicitly.\n\nThis form works in both 3.19.x and 4.0.x.\n\nChange-Id: Ie9c76a23658b46997dcb17ba3fb7f1e700eca506\n(cherry picked from commit 306a91e9e8939d75e0309e492502fa0805cc2c8d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ee8875d64bc7046b13696a90ef7a4aa54f83103c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ee8875d64bc7046b13696a90ef7a4aa54f83103c"}]},"branch":"refs/heads/stable/train"},"ca32d2146a976237cbd32be1ff755c4672f2d9df":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-24 16:00:52.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/92/684392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/92/684392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/92/684392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/92/684392/2"}}},"commit":{"parents":[{"commit":"878fa6112947d2eb022937af11dec5f415d9fa57","subject":"Use os-endpoint instead of os-url for functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/878fa6112947d2eb022937af11dec5f415d9fa57"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-24 14:55:14.000000000","tz":60},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-24 16:00:52.000000000","tz":0},"subject":"Be explicit about auth type in functional tests","message":"Be explicit about auth type in functional tests\n\nWhen using the old style of \u0027--os-url\u0027 we didn\u0027t need to be\nexplicit, it deduced the right type. And python-openstackclient\n\u003e\u003d 4.0.0 it does for --os-endpoint. But when using \u0027--os-endpoint\u0027\nwith \u003c4.0.0 we need to state \u0027admin_token\u0027 explicitly.\n\nThis form works in both 3.19.x and 4.0.x.\n\nChange-Id: Ie9c76a23658b46997dcb17ba3fb7f1e700eca506\n(cherry picked from commit 306a91e9e8939d75e0309e492502fa0805cc2c8d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ca32d2146a976237cbd32be1ff755c4672f2d9df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/ca32d2146a976237cbd32be1ff755c4672f2d9df"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
