)]}'
{"id":"openstack%2Fosc-placement~691193","triplet_id":"openstack%2Fosc-placement~master~Ibd9d1c839f820d8c8ea0ad0979d5dca9223d11dc","project":"openstack/osc-placement","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ibd9d1c839f820d8c8ea0ad0979d5dca9223d11dc","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-10-25 05:50:28.000000000","updated":"2019-12-06 01:07:48.000000000","submitted":"2019-12-06 01:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691193-1575594301249-758459c9","meta_rev_id":"607d95a901a54e9dcac99c5318ed82de76dea64d","_number":691193,"virtual_id_number":691193,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 01:05:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-12-05 19:15:52.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-12-06 00:52:42.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":1,"date":"2019-12-06 00:52:42.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 19:15:52.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-12-05 20:45:21.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-12-06 00:52:42.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-06 01:05:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddd93085f564c2038298c3c0c4386356c7230134","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-10-25 05:50:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7780fd5d136c69404b9e62805ec97c70ff8890d1","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-10-25 06:00:19.000000000","message":"Patch Set 1: Workflow-1\n\nThis can be landed between \"Phase-1: Dec 09 - Dec 13 R-22 Ussuri-1 milestone\" and \"Phase-2: Feb 10 - Feb 14 R-13 Ussuri-2 milestone\"\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html","accounts_in_message":[],"_revision_number":1},{"id":"65923465195eb2813717449cac53b625b4bc955f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 06:08:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de142be0264c43da96256b6a0138c685 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8d608eafdd4826947bcd719d4ba7c0 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6067ad50a4d34e488ebc062abd860702 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24cdc47e532b4dfc819edc84463678b9 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c35911ea6f4758a72557fb03842d68 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3df7c912c8c941f3b2d8651fb5b1f827 : SUCCESS in 5m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac5d98c03de4498aaef43a31304d1fd8 : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b98485b0dda477e88d7df1960da6a65 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a0b9d875bdf1eaff7b338255683cf4368c92ad49","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-28 20:58:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a059737f95b92403eb3f17bd1eeeed1fdf90c2e6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 22:29:21.000000000","message":"Topic changed from drop-py2 to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"6ead44417d0f8777b0cc026cb9acbc8d86bc0f8f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-13 23:01:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d5de870af97848605b61c37f716df2c0b03ff15","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-03 06:38:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3eb6df16bb06abb97f554fa8948ba485a43aaa59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 06:38:41.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/osc-placement on branch master.  The error was:\n\n  The key \"gate\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  templates:\n        - openstack-python3-ussuri-jobs\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional-py36:\n              required-projects:\n                - openstack/placement\n          - openstack-tox-functional-py37:\n              required-projects:\n                - openstack/placement\n      gate:\n        jobs:\n          - openstack-tox-functional-py36:\n              required-projects:\n                - openstack/placement\n      gate:\n        jobs:\n          - openstack-tox-functional-py37:\n              required-projects:\n                - openstack/placement\n\n  in \"openstack/osc-placement/.zuul.yaml@master\", line 2, column 5","accounts_in_message":[],"_revision_number":2},{"id":"fee721727b59ef6bc22205f395c73beed9419000","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-03 06:48:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4db745ee5570ed9dcebf511310c9013b0e0c363","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-03 06:52:33.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b085b36b19c6a20e011f11353ee5b8a1a926a654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 06:57:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b10598e691f94ddf9800ad84a1bb4134 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba232f39d74947e7b38a28a3915a40ce : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19f987a645134ed4a9a6ee400eb8832a : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60c602ab6943423a9c3b5fb3e0776d44 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc3574385934b39a87921c599f55fe3 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55dc6834ea8e44048ff2830be1e1b40a : SUCCESS in 4m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70a3deed549b40118b4b51c89e51cd1a : SUCCESS in 4m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/61cd967e2e4748d8907c2d70f183a097 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"ff5c0065e7d3af752fa4d5d66bc8ed67b6146592","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-12-05 19:15:52.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nI\u0027m okay with this except for the weird `functional` env. But I guess that\u0027s not something we can/should fix here.","accounts_in_message":[],"_revision_number":3},{"id":"57e7f1468f116f3410f9581016413893454473a1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-05 20:45:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2df5648463195170dc8b14a3d93f6472c4b00913","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-06 00:52:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)\n\nAgree with Eric that the functional environments are a bit weird but we could clean those up separately later.","accounts_in_message":[],"_revision_number":3},{"id":"146e7ddc17ed19646e1fbb6a86212b1c76586762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 00:52:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f68e0e46295b06e15c676c5a3640771f8e68ab22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:05:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"412bba638aa603f54cc72883d44b662f524579f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:05:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4146cda8d6004b9fbad59c4174973e99 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2ec9f3e93b4b5ebcbc04f2feeb1334 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30392bce99a844a2ad0b6700577ab11f : SUCCESS in 2m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abbe7f731800456486f93de21fb69d43 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6cffffaf034856807d08a77274ccab : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/116d70688dc4422ca48018b80010e04c : SUCCESS in 3m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1383fa51d544b77804e96f93e23222f : SUCCESS in 5m 29s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1fed2014f700413baaec05d309947b1a : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"607d95a901a54e9dcac99c5318ed82de76dea64d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:07:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2515058ea6248ecb75fa09d46f34fe5 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ea40611d154458483ce0694ebd89313 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"52ef27a8ce64878c2ef95298e5b9e7370e06f98a","revisions":{"109c6227b771aace20cb93e37bd58b47cb9017db":{"kind":"REWORK","_number":1,"created":"2019-10-25 05:50:28.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/93/691193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/93/691193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/1"}}},"commit":{"parents":[{"commit":"6cec98cf6f22c8ae2c5ff124b6194a162a7af882","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6cec98cf6f22c8ae2c5ff124b6194a162a7af882"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-25 05:45:09.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-25 05:49:47.000000000","tz":0},"subject":"Drop support for python2","message":"Drop support for python2\n\nThis patch stop testing with python2 and issue the release note.\n\nChange-Id: Ibd9d1c839f820d8c8ea0ad0979d5dca9223d11dc\nDepends-On: https://review.opendev.org/#/c/688997/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/109c6227b771aace20cb93e37bd58b47cb9017db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/109c6227b771aace20cb93e37bd58b47cb9017db"}]},"branch":"refs/heads/master"},"aac468dbe58cdc5631921e281b41279db0d6c6e1":{"kind":"REWORK","_number":2,"created":"2019-12-03 06:38:27.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/93/691193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/93/691193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/2"}}},"commit":{"parents":[{"commit":"6cec98cf6f22c8ae2c5ff124b6194a162a7af882","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6cec98cf6f22c8ae2c5ff124b6194a162a7af882"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-25 05:45:09.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 15:47:05.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\n\"osc-placement\" is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Ibd9d1c839f820d8c8ea0ad0979d5dca9223d11dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/aac468dbe58cdc5631921e281b41279db0d6c6e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/aac468dbe58cdc5631921e281b41279db0d6c6e1"}]},"branch":"refs/heads/master"},"52ef27a8ce64878c2ef95298e5b9e7370e06f98a":{"kind":"REWORK","_number":3,"created":"2019-12-03 06:48:43.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/93/691193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/93/691193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/93/691193/3"}}},"commit":{"parents":[{"commit":"6cec98cf6f22c8ae2c5ff124b6194a162a7af882","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/6cec98cf6f22c8ae2c5ff124b6194a162a7af882"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-10-25 05:45:09.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-11-23 15:57:43.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\n\"osc-placement\" is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Ibd9d1c839f820d8c8ea0ad0979d5dca9223d11dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/52ef27a8ce64878c2ef95298e5b9e7370e06f98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/52ef27a8ce64878c2ef95298e5b9e7370e06f98a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
