)]}'
{"id":"openstack%2Fosc-placement~691422","triplet_id":"openstack%2Fosc-placement~master~I627bfd1ff699d075028da6afafbe7fb9b2f13058","project":"openstack/osc-placement","branch":"master","topic":"story/2006779","hashtags":[],"change_id":"I627bfd1ff699d075028da6afafbe7fb9b2f13058","subject":"Add resource provider allocation unset command","status":"MERGED","created":"2019-10-25 19:34:43.000000000","updated":"2019-11-25 17:50:09.000000000","submitted":"2019-11-25 17:21:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":62,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691422-1574702485486-8d7acffc","meta_rev_id":"a68b82c6c702e6b8442eaafda7fe3a7dfa9254a9","_number":691422,"virtual_id_number":691422,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-11-20 14:33:51.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 17:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2019-11-25 17:10:33.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-11-25 10:45:14.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2019-11-25 17:10:33.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 20:39:46.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-11-25 10:45:14.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-11-25 17:10:33.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-25 17:21:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42b3188330f7fbf2c0a6f54c415492077f30506a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-25 19:34:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43d8d4bd61847c1fd5792667cdc1350917e290cb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-25 19:36:53.000000000","message":"Patch Set 1: Workflow-1\n\nNeed to add that 1.8 test.","accounts_in_message":[],"_revision_number":1},{"id":"aa7026df9cac4dc1290d774268e2597922021fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 19:43:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3222e1fe2586463fb3ac7fcd2741a23b : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8438618e41224e1d92e19171f79af65b : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14d071ee738f4a4a9db860d3faad8945 : SUCCESS in 2m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c613f80cad6402db8e37bfd4ddb7654 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306435a7d63e499b9a093025c2852e94 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8faa4ea302474e27ad321ab76a363ec4 : SUCCESS in 2m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3dfa259e6f7c48439d525c4869ac3a80 : SUCCESS in 3m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/07aeddbd2d0e4972836b2ff5e2db600d : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"443ad5e7558e404e6932ee3010d63482de797141","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-28 20:50:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7498b6290bf9bf4f9b937c62e4504c16c8356cc6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-28 20:55:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"38fd1fd5ce7cad042ac077d6fee961f50d812c5f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-28 20:55:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"392c5d035d841183494d2ead2acdbba6f4a01705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 21:10:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b3d534e646c42cfbf380a3516e6dff7 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77660b0bafc8460ba816b8f55cfc4a13 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74cc7f2a1d034e62a245d9f4db0e4571 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c0b4cb885804080b5fea2530e6125d4 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada8525fc3804590badce00c5e6466d9 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/883805d1cd9d41c693069571e0b6f86c : SUCCESS in 2m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5a732bf3bd0c4ff282176f35910385e3 : SUCCESS in 3m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6800d428bc8c45afa957b65184ee517a : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e7dfd7eb8e2d8417e01a2ffef051736133d47289","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-28 21:12:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f13a92f0c5c37f26d34a58c11e6081a591c91429","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-04 01:08:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0303dc42f9d99d9a79690f7de3a9d5481302dece","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-13 20:39:46.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"586696d0739f582e6b263a5f3908ab707f230eab","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-13 20:55:55.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7b0bf08b7b5c5a48e46b6ca7fa6cdbdabde05bca","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-14 10:02:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6051f933bdbea25f6c810399cb810752d7082dd8","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-14 10:11:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1130c15512c2a5b5580183f3b18321134e913d4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-14 20:48:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32dc299f5fbc23e6f8d6ddf4005ac62e8e27ddca","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-14 21:54:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"02af08a566b2218eba215d6b220343f13e931155","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-14 21:56:45.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4cfc45d6ee67c0c2131a0aae16f7eec03b64999b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 22:05:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d13bb397f3a467681803a00d8c552b2 : SUCCESS in 2m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/229394841e5248d2be7ce3c0bf02cc6b : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/945203b6d2754b46958a5e2ad5818e45 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f74d46a4cad743baa43f8ac6a0625874 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d500c1ea48754ca4b9f9ce05d0f77002 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b960b9a760ff46c9b602c15760c79730 : SUCCESS in 2m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4e23ccf11a274131a67111b93cf656ed : SUCCESS in 4m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be90ceff9b604dc7b96efdedb1399a6f : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"c487729f77ad2b23f1090809ce5c690d695f07c7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-14 23:19:41.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nI\u0027m sure there was already plenty of bikeshedding to come up with \"--no-provider\" as a name for this option, so I won\u0027t downvote on it, but coming into this fresh, I don\u0027t like it at all. Even in coordination with other --no-* opts, it more implies that we\u0027re either getting rid of or ignoring the provider than that we\u0027re specifically targeting its allocations. There must be a better option. \"--clear-provider\"? \"--empty-provider\"?\n\nHowever, I am going to downvote for the missing consumer generation, at least until that behavior is validated in a live environment. If we want to defer support for 1.28+ that\u0027s one thing, but then we shouldn\u0027t recommend using it.","accounts_in_message":[],"_revision_number":4},{"id":"7387dd9ded6ba0da0d7bc2bca0d14c0465c9cc34","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 14:12:43.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e I\u0027m sure there was already plenty of bikeshedding to come up with \"--no-provider\" as a name for this option, so I won\u0027t downvote on it, but coming into this fresh, I don\u0027t like it at all. Even in coordination with other --no-* opts, it more implies that we\u0027re either getting rid of or ignoring the provider than that we\u0027re specifically targeting its allocations. There must be a better option. \"--clear-provider\"? \"--empty-provider\"?\n\nI have some notes in the linked story about options. I originally wanted to do \"openstack resource provider allocation delete --provider \u003cprovider_id\u003e \u003cconsumer_id\u003e\" but the delete CLI doesn\u0027t print anything out like the set CLI does and I thought changing that would be weirder than updating set since I think the primary use case for this change is removing one or more but not all providers from the allocations for a given consumer. If you want to remove them all, you\u0027d use the allocation delete CLI. We handle DELETE here just in case the caller is confused or loses a race or something.\n\nAnother option I\u0027ve kicked around is adding a new \"openstack resource provider allocation unset --provider \u003cprovider_id\u003e \u003cconsumer_id\u003e\" command. If we did that, I think --provider would be required and I don\u0027t think I\u0027d want to get into the mess of adding a --no-allocation to an unset command (the OSC pattern seems to be set --thing, unset --no-thing). So what do people think about adding an unset command where one or more provider is required rather than munging this into the set command where either --allocation or --no-provider is required?\n\nOr alternatively change the name on this option to be more clear, e.g. --remove-provider, --clear-provider, etc? I don\u0027t know how Dean would feel about that so I\u0027d want to ask him.\n\n\u003e However, I am going to downvote for the missing consumer generation, at least until that behavior is validated in a live environment. If we want to defer support for 1.28+ that\u0027s one thing, but then we shouldn\u0027t recommend using it.\n\nI will add a test for 1.28 but I think it\u0027s going to be OK because GET and PUT are mirroring the consumer_generation parameter.","accounts_in_message":[],"_revision_number":4},{"id":"a87d69793ee4f60d3e28bc2805603a4fb46e8538","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 14:32:33.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29616eaa11436e0e35a8a7e87698f9200545d753","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 14:35:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd5170bb9d4cc41acf662230bb6a96fca708853f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 14:57:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"469b82d1df528f586b9fa4dfd3a9f4b99c769f93","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 15:03:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ef9a048117e691e912c578687e968f7f1c993264","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 15:08:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ff68bd7b048323383a8ac18a479c1170821d062c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 15:11:13.000000000","message":"Patch Set 5:\n\nOK PS5 handles the consumer_generation stuff. I\u0027m posting it just so y\u0027all can see the difference.\n\nTwo questions remain:\n\n1. Should I split out the 1.28 consumer_generation stuff into a separate change before this? It\u0027s probably easy enough to do.\n\n2. Naming; should we:\n\na) continue to use the set command but change the name of the option to --clear-provider or --remove-provider or\n\nb) use a new unset command with a --provider option\n\nI\u0027m a bit inclined to (a) since then I wouldn\u0027t have to duplicate the consumer generation and a lot of the code between the commands and also side-step questions about if an unset command should have a --no-allocation option or something that allows removing specific resource classes from an existing allocation (that would be annoying to deal with on a CLI).","accounts_in_message":[],"_revision_number":5},{"id":"1da897299e2b10c263d3a99f89a1d82de4e8ebd5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 15:17:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bc0e09ae723c3a08b0fac93206c722e453e2a515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 15:18:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0a9dc66b294bd2a0ebc1b2ceddf69a : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbbe451b6df3456ea5f7f9ccc99f9c61 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8f6bf03fea7441994930e6475812ff1 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1962234f9804489b57ccd7be0a26ff0 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d8a7efd37e4e71b1072391aca9e644 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2caf6bcc3ef04135951e9e0ad2338107 : SUCCESS in 3m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2f1d8afc92da4826a894048c26b0143a : SUCCESS in 3m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49918e4a268041c9a4186b72154783ce : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":5},{"id":"6103dac23fa9617fefe0863112d26bcf451077bd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-15 15:57:43.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nIn reverse order:\n\n\u003e 2. Naming; should we:\n\u003e a) continue to use the set command but change the name of the option to --clear-provider or --remove-provider or\n\u003e b) use a new unset command with a --provider option\n\nI prefer some variant of the latter. Have we sullied \u0027set\u0027 with any other partial-update stuff yet? If not, I would say we should try to leave \u0027set\u0027 as full overwrite and put partial updates in a different subcommand.\n\n\u003e duplicate the consumer generation and a lot of the code between the commands\n\nHelper methods\n\n\u003e if an unset command should have a --no-allocation option\n\nIf it\u0027s unset, you don\u0027t really need \u0027no-\u0027 IMO. Something like\n\n openstack resource provider allocation unset [--provider P] [--resource-class R1 [--resource-class R2 ...]] $consumer\n\nWith none of the optional args, it\u0027s basically the same as delete. Otherwise you\u0027re narrowing to a specific provider and/or resource class(es).\n\nAnother option is a patch-like command (resource provider allocation update?) that lets you add/remove pieces granularly. But that would be a bear to design -- just use `set`.\n\n\u003e 1. Should I split out the 1.28 consumer_generation stuff into a separate change before this? It\u0027s probably easy enough to do.\n\nIf you\u0027re doing a new subcommand, yes; if you can talk me into leaving this stuff on `set`, nah, just leave it folded in.","accounts_in_message":[],"_revision_number":5},{"id":"60d0ec9c2bccafa942efffefbf69bbd6f059a9e6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-15 18:16:01.000000000","message":"Patch Set 5: Code-Review-1\n\nIn IRC [1] we agreed on \n\n openstack resource provider allocation unset [--provider $rp_uuid ...] $consumer_uuid\n\nwhere --provider is repeatable, and we can later add (repeatable) --resource-class $rcname. The effect is limited to the provider(s) (and resource class(es)) specified, or all if you don\u0027t specify any.\n\nThat being the case, it would be nice to add 1.28 support for `set` in a separate patch. Though it is of questionable value to the end user because of the shortness of the window between GET and PUT, today it will simply fail.\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-sdks/%23openstack-sdks.2019-11-15.log.html#t2019-11-15T14:13:27","accounts_in_message":[],"_revision_number":5},{"id":"9d3b291c7c956bc405f76de956416ae484024ff1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-15 19:16:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"148f181f2118e0b2e96e5e5c1c5bbdadb9b97119","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-19 17:09:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2c47c6732c88692921cfe321bf92ce38264d3ccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 17:19:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/05939e4702a343eb92a2d9ce61361ec6 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ec3b2c94304a6d9961bf06e20e3abf : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/edfbd1f03d4a4bed83b2c9f1468e20f5 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e8fd220af2e4b6c9602bd104bc510e6 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d2318b36edb4872b017986953b855b1 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86d72c1446b4e40824c4782032a98bf : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72f1d21f3ce24180b6282753ca58dbd1 : SUCCESS in 4m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a80fdaad2e1a4dab96f77cf024f3842c : SUCCESS in 4m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65db57bf972243059370d456de69469e : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":6},{"id":"383889e891363c1a89d059c245fd86ded2fc9d6a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-19 17:37:42.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a01286688823f7f43c3a8260b34a9b2a7c93be57","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-19 17:45:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9fc70cf4ebcca3bf403892cfe7f589815aea63d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 17:57:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/46abc445ca9941f3bb73e46750149edd : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58659dd7be1e48d98bbbb9ca6314866f : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e970b5a66f594d98af3dbfec7a1d6d07 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb11cb2989b74beeb3629b6623f4be71 : SUCCESS in 2m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5766d1cf29f74aa08c8ef8ed74853bf2 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a699f4482b1425297dfb9b545d052fd : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ef48c8de13b4c3c9ebaf4a2764a2dbf : SUCCESS in 3m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/56a4edb57e604017a34ae4654b4b3579 : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f292986eef94e97845bf691c1440621 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":7},{"id":"dd1d2d22e31c0163ffed9c243eb77bc4905cab68","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-20 13:10:44.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nThank you for updating. Some comments within.","accounts_in_message":[],"_revision_number":7},{"id":"e05e4f0d4adff9421897936373dd73913eca479e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-20 14:33:51.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a78466cc0bd91c005d57da7f1a668270c8ea7d1d","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-21 11:35:25.000000000","message":"Patch Set 7: -Code-Review\n\n(3 comments)\n\nRemoved my -1  because the gap seems to be acceptable.\nI\u0027ll revisit this after the next one.","accounts_in_message":[],"_revision_number":7},{"id":"6c8c86ac1459866b2848ab9b0896eadd421fccb9","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-25 10:45:14.000000000","message":"Patch Set 7: Code-Review+2\n\nThe next patch looks good to me as well.","accounts_in_message":[],"_revision_number":7},{"id":"6ac845ec17d571a45c0f256598dcc8bf2b41c6f9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-25 17:10:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(6 comments)\n\nThis wfm. Added a couple of things for your FUP.","accounts_in_message":[],"_revision_number":7},{"id":"737a7d07f77a0110fc2a4c4981deaca157111efa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:10:44.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"db152a6cb7eb11798a7767000095e180110c20e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:21:25.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a06627ba29dd411784eef8cd1be6f179 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66cfcdd27d0c4d1ab5887d6e9949c58f : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ed33891d76f4c5e9d4d5cef6306a89a : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cd40bfd2a754b63b0019fbbf1be9e57 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b463490bc3c2444cbdae792ddc12429c : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef07bdc912d4d178fa675954dd9edff : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07b086b08a3c4bfeabdd31ca98da5b0d : SUCCESS in 5m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/49ebe855e73f44ddaf3a723586bc19d4 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/858b8d8f5f5a4d8e85d6cab0b34f6997 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":7},{"id":"5b1499392dbbe1d6e4f66dc310e55230029f6f1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:21:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"5a010059ab3208af1cc1bcf0b4ed7682818a8aca","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-25 17:23:04.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"abeedf2e02b5e1bd5e8c8ee8515a0001431c1b30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:25:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ddfc129d464af7acf52c8624ea4013 : SUCCESS in 1m 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3caa0f17cc1a4be8900dffc2bd022e4a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7},{"id":"faaa2e135d811249c3f5e7b7652f4c307345bf3f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-25 17:30:26.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"66563df3eb1157fa6ac856d74fee383182847ee4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-25 17:50:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9b86d945594bd22980dcf76ebcb8fb5112d471ed","revisions":{"5781572d2941b587830abd108ac71af5b115751e":{"kind":"REWORK","_number":1,"created":"2019-10-25 19:34:43.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/22/691422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/22/691422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/1"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-25 19:33:30.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-25 19:33:30.000000000","tz":-240},"subject":"Add --no-provider option to resource provider allocation set","message":"Add --no-provider option to resource provider allocation set\n\nA new ``--no-provider`` option is added to the\n``openstack resource provider allocation set`` command which is mutually\nexcluse with the ``--allocation`` option and will allow removing a specific\nresource provider from the set of allocations for the given consumer.\nThis can be useful when a consumer has allocations against more than one\nresource provider and ``openstack resource provider allocation delete``\nis undesirable as it removes all allocations for the consumer.\n\nChange-Id: I627bfd1ff699d075028da6afafbe7fb9b2f13058\nStory: #2006779\nTask: #37304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5781572d2941b587830abd108ac71af5b115751e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/5781572d2941b587830abd108ac71af5b115751e"}]},"branch":"refs/heads/master"},"50af508ce75790a55c7570b51bf3bf1763491bfb":{"kind":"REWORK","_number":2,"created":"2019-10-28 20:50:20.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/22/691422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/22/691422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/2"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-25 19:33:30.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-28 20:49:09.000000000","tz":-240},"subject":"Add --no-provider option to resource provider allocation set","message":"Add --no-provider option to resource provider allocation set\n\nA new ``--no-provider`` option is added to the\n``openstack resource provider allocation set`` command which is mutually\nexcluse with the ``--allocation`` option and will allow removing a specific\nresource provider from the set of allocations for the given consumer.\nThis can be useful when a consumer has allocations against more than one\nresource provider and ``openstack resource provider allocation delete``\nis undesirable as it removes all allocations for the consumer.\n\nChange-Id: I627bfd1ff699d075028da6afafbe7fb9b2f13058\nStory: #2006779\nTask: #37304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/50af508ce75790a55c7570b51bf3bf1763491bfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/50af508ce75790a55c7570b51bf3bf1763491bfb"}]},"branch":"refs/heads/master"},"c1474f979d00031f3f864fb6ac95082957bce357":{"kind":"REWORK","_number":3,"created":"2019-10-28 20:55:56.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/22/691422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/22/691422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/3"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-25 19:33:30.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-28 20:55:38.000000000","tz":-240},"subject":"Add --no-provider option to resource provider allocation set","message":"Add --no-provider option to resource provider allocation set\n\nA new ``--no-provider`` option is added to the\n``openstack resource provider allocation set`` command which is mutually\nexcluse with the ``--allocation`` option and will allow removing a specific\nresource provider from the set of allocations for the given consumer.\nThis can be useful when a consumer has allocations against more than one\nresource provider and ``openstack resource provider allocation delete``\nis undesirable as it removes all allocations for the consumer.\n\nChange-Id: I627bfd1ff699d075028da6afafbe7fb9b2f13058\nStory: #2006779\nTask: #37304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c1474f979d00031f3f864fb6ac95082957bce357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c1474f979d00031f3f864fb6ac95082957bce357"}]},"branch":"refs/heads/master"},"9b2f9891dbee65dcd7e1c9205b762cc316c5abc5":{"kind":"REWORK","_number":4,"created":"2019-11-14 21:54:37.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/22/691422/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/22/691422/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/4"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-25 19:33:30.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-11-14 21:53:27.000000000","tz":-300},"subject":"Add --no-provider option to resource provider allocation set","message":"Add --no-provider option to resource provider allocation set\n\nA new ``--no-provider`` option is added to the\n``openstack resource provider allocation set`` command which is mutually\nexclusive with the ``--allocation`` option and will allow removing\nreource providers from the set of allocations for the given consumer.\nThis can be useful when a consumer has allocations against more than one\nresource provider and ``openstack resource provider allocation delete``\nis undesirable as it removes all allocations for the consumer.\n\nNote that --no-provider can be specified multiple times.\n\nAlso, if the resulting allocations are all gone, a DELETE is called rather\nthan PUT since PUT /allocations/{consumer_id} does not allow an empty\n\"allocations\" dict in the request body. In this case nothing is output\nsimilar to the \"openstack resource provider allocation delete\" command.\n\nChange-Id: I627bfd1ff699d075028da6afafbe7fb9b2f13058\nStory: #2006779\nTask: #37304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9b2f9891dbee65dcd7e1c9205b762cc316c5abc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9b2f9891dbee65dcd7e1c9205b762cc316c5abc5"}]},"branch":"refs/heads/master"},"c7e51861af56152374523d497e73f8441ca236dd":{"kind":"REWORK","_number":5,"created":"2019-11-15 15:08:37.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/22/691422/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/22/691422/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/5"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-25 19:33:30.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-11-15 15:07:41.000000000","tz":-300},"subject":"Add --no-provider option to resource provider allocation set","message":"Add --no-provider option to resource provider allocation set\n\nA new ``--no-provider`` option is added to the\n``openstack resource provider allocation set`` command which is mutually\nexclusive with the ``--allocation`` option and will allow removing\nreource providers from the set of allocations for the given consumer.\nThis can be useful when a consumer has allocations against more than one\nresource provider and ``openstack resource provider allocation delete``\nis undesirable as it removes all allocations for the consumer.\n\nNote that --no-provider can be specified multiple times.\n\nIf the resulting allocations are all gone and 1.28 or greater is not\nbeing used, a DELETE is called rather than PUT since\nPUT /allocations/{consumer_id} before 1.28 does not allow an empty\n\"allocations\" dict in the request body. In this case nothing is output\nsimilar to the \"openstack resource provider allocation delete\" command.\n\nChange-Id: I627bfd1ff699d075028da6afafbe7fb9b2f13058\nStory: #2006779\nTask: #37304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c7e51861af56152374523d497e73f8441ca236dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/c7e51861af56152374523d497e73f8441ca236dd"}]},"branch":"refs/heads/master"},"151321698a1c8897031d1365846b65d2ef8fb9e5":{"kind":"REWORK","_number":6,"created":"2019-11-19 17:09:47.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/22/691422/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/22/691422/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/6"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-25 19:33:30.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-11-19 17:09:38.000000000","tz":-300},"subject":"Add resource provider allocation unset command","message":"Add resource provider allocation unset command\n\nAdds the \"openstack resource provider allocation unset\" command which\nallows removing resource provider(s) from the set of allocations for\nthe given consumer. This can be useful when a consumer has allocations\nagainst more than one resource provider and\n\"openstack resource provider allocation delete\" is undesirable as it\nremoves all allocations for the consumer. The new unset command requires\n\"--os-placement-api-version 1.12\" or greater.\n\nNote that the --provider option can be specified multiple times.\n\nIf the resulting allocations are all gone and 1.28 or greater is not\nbeing used, a DELETE is called rather than PUT since\nPUT /allocations/{consumer_id} before 1.28 does not allow an empty\n\"allocations\" dict in the request body. In this case nothing is output\nsimilar to the \"openstack resource provider allocation delete\" command.\n\nChange-Id: I627bfd1ff699d075028da6afafbe7fb9b2f13058\nStory: #2006779\nTask: #37304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/151321698a1c8897031d1365846b65d2ef8fb9e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/151321698a1c8897031d1365846b65d2ef8fb9e5"}]},"branch":"refs/heads/master"},"9b86d945594bd22980dcf76ebcb8fb5112d471ed":{"kind":"REWORK","_number":7,"created":"2019-11-19 17:45:52.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/22/691422/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/22/691422/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/22/691422/7"}}},"commit":{"parents":[{"commit":"3ac639bcf766bc05c1899ca5cec91e415d4f8456","subject":"Fix deps for using venv tox target to create a release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/3ac639bcf766bc05c1899ca5cec91e415d4f8456"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-25 19:33:30.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-11-19 17:38:42.000000000","tz":-300},"subject":"Add resource provider allocation unset command","message":"Add resource provider allocation unset command\n\nAdds the \"openstack resource provider allocation unset\" command which\nallows removing resource provider(s) from the set of allocations for\nthe given consumer. This can be useful when a consumer has allocations\nagainst more than one resource provider and\n\"openstack resource provider allocation delete\" is undesirable as it\nremoves all allocations for the consumer. The new unset command requires\n\"--os-placement-api-version 1.12\" or greater.\n\nNote that the --provider option can be specified multiple times.\n\nIf the resulting allocations are all gone and 1.28 or greater is not\nbeing used, a DELETE is called rather than PUT since\nPUT /allocations/{consumer_id} before 1.28 does not allow an empty\n\"allocations\" dict in the request body. In this case nothing is output\nsimilar to the \"openstack resource provider allocation delete\" command.\n\nChange-Id: I627bfd1ff699d075028da6afafbe7fb9b2f13058\nStory: #2006779\nTask: #37304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9b86d945594bd22980dcf76ebcb8fb5112d471ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9b86d945594bd22980dcf76ebcb8fb5112d471ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
