)]}'
{"id":"openstack%2Fosc-placement~695048","triplet_id":"openstack%2Fosc-placement~master~I8f85ee4ff4c39fba69a78bbe642f79a706504df8","project":"openstack/osc-placement","branch":"master","topic":"story/2006779","hashtags":[],"change_id":"I8f85ee4ff4c39fba69a78bbe642f79a706504df8","subject":"Add support for microversion 1.28 in allocation set","status":"MERGED","created":"2019-11-19 17:45:52.000000000","updated":"2019-11-25 17:34:42.000000000","submitted":"2019-11-25 17:31:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695048-1574703117074-2b47506d","meta_rev_id":"70ce8e157c5a9934e69af018b2c82cbabc60c1f6","_number":695048,"virtual_id_number":695048,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-11-21 14:26:55.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 17:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-11-25 17:10:46.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-11-25 10:55:18.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-11-25 17:10:46.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 10:55:18.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-11-25 17:10:46.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-25 17:31:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c308e2c7465fb4778e5590c1a1bfefc645448cf8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-19 17:45:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b70ecf2d5a8a1db87fe1e45195065922aa476e2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 18:07:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1813e3def4924addbbab964a4008bc49 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/936877744ce847f1a307b5fff88df440 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e28fe518d94a4a99bd0ad449f9925c05 : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35687c97d4754462a426ecc8016544ff : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9751a24d8bb34ee58aab4a9c8c7d5ed1 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/007cfdb39aec4eb5a621d278a132780e : SUCCESS in 3m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2be7d7053c7141328f35fe9c3150f39b : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc40a629cb244773a8da9163f2d52b92 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"287b03b361d0a6e10616248bc828da10e401db47","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-21 12:12:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nThanks. LGTM overall.\nJust one point on L149: Don\u0027t we want to display the updated consumer generation?","accounts_in_message":[],"_revision_number":1},{"id":"2cb2f8406440cc90165690671b7d83182ebbaccf","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-21 12:17:25.000000000","message":"Patch Set 1:\n\nThis is just a note… we have 4 APIs changed in 1.28\n\nPUT /allocations/{consumer_uuid} -\u003e  Done in this (and the previous) patch\nPOST /allocations -\u003e API not implemented in client \nGET /allocations/{consumer_uuid} -\u003e Can be done in this or another patch later.\nGET /resource_providers/{uuid}/allocations -\u003e Done without any changes (see the following)\n\n$ openstack resource provider show $test1 --allocations --os-placement-api-version\u003d1.28\n+----------------------+----------------------------------------------------------------------------------------------------+\n| Field                | Value                                                                                              |\n+----------------------+----------------------------------------------------------------------------------------------------+\n| uuid                 | b52ae3a3-d70b-4b03-a21b-163a0c732aff                                                               |\n| name                 | test1                                                                                              |\n| generation           | 4                                                                                                  |\n| root_provider_uuid   | b52ae3a3-d70b-4b03-a21b-163a0c732aff                                                               |\n| parent_provider_uuid | None                                                                                               |\n| allocations          | {u\u00278b375eda-a233-4c31-96fe-b13f0021ea31\u0027: {u\u0027consumer_generation\u0027: 3, u\u0027resources\u0027: {u\u0027VCPU\u0027: 4}}} |\n+----------------------+----------------------------------------------------------------------------------------------------+","accounts_in_message":[],"_revision_number":1},{"id":"eefd1873c286ee24166b8d7c1f7f8948ef9261dd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-21 14:25:13.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0cf1d6e53689ee652df76d209fec4ff25a3e875c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-21 14:26:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68b542f1ce54449a33c37541be57f838ee8e392a","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-22 01:23:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e566194a9e6b9fc15c02ec4a0e880288ca72799","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-22 06:24:40.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cc8126fcd198d9ce35975ca7a98db77898bfd0e","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-11-25 10:55:18.000000000","message":"Patch Set 1: Code-Review+2\n\nIt looks like only ``openstack resource provider aggregate set`` accepts the resource provider generation and we could expose resource provider and/or consumer generation for following APIs, \n\n* openstack resource provider inventory set\n* openstack resource provider allocation set\n\n...but since whether to use generation or not can be a user\u0027s choice, we can do it later optionally if more strict handling is needed, so this looks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"ffdfae0190b80d9f5d85d7d2d99b9c8f3b444cd4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-25 16:39:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nI looked at these out of order.\n\nThis one looks okay to me, but let me come back to it after looking at the predecessor.","accounts_in_message":[],"_revision_number":1},{"id":"e3175ea4de476b353b4764a0f4f17cc9cfa6c76e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-25 17:10:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\npredecessor is good, let\u0027s go.","accounts_in_message":[],"_revision_number":1},{"id":"04fa85c3743c12fa4803d2298a0702688ef94574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:10:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3eae928f039901b7b8f95ac09746d8885378a17d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:31:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be4ffb23771540a5b5ae2fbfe9c908f8 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50cbf0aa462547f6b830276e3d567e32 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb04a28117340639e432af3e9d3fa5d : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06a7e1b6108845af8d0b63d022053e69 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a0ced489ff4501b56842607d263282 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c4518668cd04ace9f1b83913aae3289 : SUCCESS in 3m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/094b348ff0974c188c94a8e0b9227d23 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87bd64ca572e4e8ca8478e71f1b6fca3 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"00991c063458f040b51b6458aad1ed69fc26cf6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:31:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5be8ce6434402959c6b9e5baef34cca8d5e3e0ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:34:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028747272977484aaa4511da6de7d625 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ae4f24b9dc04a5980cd2ae67119bd5e : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0acbe401c57ea930293d3c6bef7d0a9d54bcbf4d","revisions":{"0acbe401c57ea930293d3c6bef7d0a9d54bcbf4d":{"kind":"REWORK","_number":1,"created":"2019-11-19 17:45:52.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/48/695048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osc-placement","ref":"refs/changes/48/695048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/48/695048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/48/695048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osc-placement refs/changes/48/695048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osc-placement refs/changes/48/695048/1"}}},"commit":{"parents":[{"commit":"9b86d945594bd22980dcf76ebcb8fb5112d471ed","subject":"Add resource provider allocation unset command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/9b86d945594bd22980dcf76ebcb8fb5112d471ed"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-11-19 17:41:38.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-11-19 17:41:38.000000000","tz":-300},"subject":"Add support for microversion 1.28 in allocation set","message":"Add support for microversion 1.28 in allocation set\n\nThis adds support to the resource provider allocation set command\nfor microversion 1.28 which uses consumer generations to ensure\natomic updates of consumer allocations.\n\nChange-Id: I8f85ee4ff4c39fba69a78bbe642f79a706504df8\nStory: #2006912\nTask: #37557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0acbe401c57ea930293d3c6bef7d0a9d54bcbf4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osc-placement/commit/0acbe401c57ea930293d3c6bef7d0a9d54bcbf4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
